)]}'
{"id":"openstack%2Frally~923149","triplet_id":"openstack%2Frally~master~I572adaa518ffac6113115360a2d8030c7e405b4b","project":"openstack/rally","branch":"master","topic":"drop-centos-8-stream","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-07-03 08:55:01.000000000","reason":"Change was submitted"},"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2024-07-03 08:39:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I572adaa518ffac6113115360a2d8030c7e405b4b","subject":"Drop centos 8 stream jobs","status":"MERGED","created":"2024-07-01 11:30:44.000000000","updated":"2024-07-03 08:55:01.000000000","submitted":"2024-07-03 08:55:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923149-drop-centos-8-stream","meta_rev_id":"0e0459303e7fb3c1dabc1156e596c73ae46939a7","_number":923149,"virtual_id_number":923149,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 08:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2024-07-03 08:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2024-07-03 08:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-01 11:47:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 08:26:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-03 08:26:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"a1f583a50a8f7dc1ab55145d65b84e135dd0f0a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-07-01 11:30:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29c2ce03d4e38d3510e0270f4b545eb6b9e361ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 11:47:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a58a93183d4125aa17b9d8dc192e6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42731c7e37ab4aabac05374e7bcf49dc : SUCCESS in 4m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/4b2b0fcf5a4744e3a296d70d5e73fefc : SUCCESS in 4m 04s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d6b84736bad444afbb87b3769ef4c768 : SUCCESS in 3m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0aaaf0597f48359797e3a568531665 : SUCCESS in 2m 39s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/86ae1f3d40814d73a736126bfef278f0 : SUCCESS in 4m 17s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/bfc98437e5de4c0a979437c3998a3138 : SUCCESS in 2m 57s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9c7d7539394e4159a78a316bb2706464 : SUCCESS in 2m 23s\n- rally-tox-py311-sqlalchemy14 https://zuul.opendev.org/t/openstack/build/5839be664ab94a9dac765fa5d8c56279 : SUCCESS in 5m 10s\n- rally-tox-py311-sqlalchemy2 https://zuul.opendev.org/t/openstack/build/11578da9f7e64cf694b0268a8bc7a7b7 : SUCCESS in 3m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/aacae2558fa44c6f9e401aaf82365c67 : SUCCESS in 2m 15s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/bb9a6b0c53cf4ddf9e32d2e6e3d6cff8 : SUCCESS in 7m 49s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/bc4d5ae297044acdb4fcd385896fdcd0 : SUCCESS in 6m 03s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaaeb548c94a4c46a3d18f0fff184a8a : SUCCESS in 5m 15s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa8e9b4a51074f95bd50f2cd1618f2db : SUCCESS in 5m 42s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/79a365f504e443da8e8819fb8db622af : SUCCESS in 6m 11s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/5abe058e55624be89966cd3988207041 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"19a89d2a820a00eeaaae17e2e5d76c37138ad48a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-07-03 08:39:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baf1cdbdfe82a4e10bde1e156e5dfb00257f1efe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"720f996573fc1e75d6545a4d2b55a4d42d85c3a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:55:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bebed791a4246b6849a87b4e2c28cfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de48300f8abd4035ab2b6aaa7dfc1a85 : SUCCESS in 3m 37s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/13dba5e064274885a7ff120c5041ad89 : SUCCESS in 4m 16s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9f1b776a4cd74a27863434d6e39cc216 : SUCCESS in 3m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aabb2e8b343496ea972f19936566e25 : SUCCESS in 2m 36s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/1e401b0339224e57a525ea06c6bbd9cd : SUCCESS in 2m 48s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/d2be0a8409d247d99988bea6259f84c2 : SUCCESS in 4m 13s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/98af960cfa284d25b79b43499a094530 : SUCCESS in 2m 46s\n- rally-tox-py311-sqlalchemy14 https://zuul.opendev.org/t/openstack/build/c48d291da0f844fdadd7c4b0dc78dc19 : SUCCESS in 3m 29s\n- rally-tox-py311-sqlalchemy2 https://zuul.opendev.org/t/openstack/build/d0184d1619724cc98239d181ddb964e6 : SUCCESS in 3m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8299b46264db4b0387337a4e1aa2ab52 : SUCCESS in 5m 51s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ffbedc618dea4afeaea23999e4f7d826 : SUCCESS in 6m 11s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6d433187ac2485d902c613f60fb2167 : SUCCESS in 5m 31s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8bfd587903a46cc920e7a1b571f5dcb : SUCCESS in 6m 03s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/007e22fcf43b4527ba322598a0986bc8 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0e0459303e7fb3c1dabc1156e596c73ae46939a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:55:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2cd26a4cfd053b9a70b64ae79eb87718591aa63","revisions":{"b2cd26a4cfd053b9a70b64ae79eb87718591aa63":{"kind":"REWORK","_number":1,"created":"2024-07-01 11:30:44.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/49/923149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/49/923149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/49/923149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/49/923149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/49/923149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/49/923149/1"}}},"commit":{"parents":[{"commit":"094f0188fc3511ec5cf54d9e38e4f8da6c37fb1a","subject":"Propose 4.1.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/094f0188fc3511ec5cf54d9e38e4f8da6c37fb1a"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-07-01 11:29:36.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-07-01 11:30:43.000000000","tz":120},"subject":"Drop centos 8 stream jobs","message":"Drop centos 8 stream jobs\n\nCentOS 8 Stream has reached its end of life and the upstream\npackage repos have been removed. The CentOS nodes from Zuul and\nNodepool are being removed. See the following email announcement:\n\nhttps: //lists.opendev.org/archives/list/service-announce@lists.opendev.org/thread/YHPRWKWULKJATAJZPFIZCBHS7HZJE5V2/\nChange-Id: I572adaa518ffac6113115360a2d8030c7e405b4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b2cd26a4cfd053b9a70b64ae79eb87718591aa63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b2cd26a4cfd053b9a70b64ae79eb87718591aa63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
