)]}'
{"id":"openstack%2Frally~924741","triplet_id":"openstack%2Frally~master~I5909ad0222e1ce388ec7593fadbb04cb5e4a0ae3","project":"openstack/rally","branch":"master","topic":"py314","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-11-20 19:55:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5909ad0222e1ce388ec7593fadbb04cb5e4a0ae3","subject":"Python 3.14: do not use deprecated ast.Str","status":"MERGED","created":"2024-07-23 14:54:00.000000000","updated":"2024-11-20 19:55:49.000000000","submitted":"2024-11-20 19:55:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924741-py314","meta_rev_id":"a536fe9b84a8e68c8950797c268708b2c7265c16","_number":924741,"virtual_id_number":924741,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 19:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2024-11-20 19:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2024-11-20 19:30:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 15:09:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 19:30:31.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"311004af9291355994263b9a543c196f9b32f2ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-23 14:54:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"650d0767d47bcd9c5208abd5a5d43fa92efc26ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-23 14:54:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7c7185a2d10211e0b22c0e2732f030c452efeb19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 15:09:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a95b0e8958549b6801d08f2ecc3e667\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1107b396a6414d9899d76f1c2b626f67 : SUCCESS in 6m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/4a64a142325e49cba567eadc0662c7e2 : SUCCESS in 5m 25s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c71d3b19b6dd4b53a971f0b861bc2b06 : SUCCESS in 4m 34s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a0df3d547e4ee380ab8fd6e790ca00 : SUCCESS in 2m 43s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/25f18a9a855a44e4af2641ccf974a1d3 : SUCCESS in 2m 57s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/83ce77d3371b4cfe8d0d2c947a8046e0 : SUCCESS in 5m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/fa3dc96ad0b4464abfc0896cb12f840b : SUCCESS in 3m 00s\n- rally-tox-py311-sqlalchemy14 https://zuul.opendev.org/t/openstack/build/da46a41e223c4431b3cc1202e3f5611d : SUCCESS in 2m 46s\n- rally-tox-py311-sqlalchemy2 https://zuul.opendev.org/t/openstack/build/b1481516d7f74084be154aadb03422d7 : SUCCESS in 4m 24s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/3cd95cb62b7d490ebc155492b03d1ce4 : SUCCESS in 2m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/22e3583252904de7806e1985f5cfa6a8 : SUCCESS in 5m 46s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f4dcf061079644dea79078426218f36e : SUCCESS in 6m 01s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72cfa2ddd6a341d3b30ab1455ee3a3ae : SUCCESS in 5m 59s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3e0a960b2104b3daee540ee0827c411 : SUCCESS in 5m 52s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/92c57f4298ab4e54961c5fa482f4f26f : SUCCESS in 6m 19s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e8cbc6396e1246f89d86734a9926a083 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d187a23f32296182b7d771a8871b1a98e148b7ef","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-11-20 19:30:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"531d11609aefd38efeed1f8f2bb3d108b7e7f99c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:30:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79c93ddda7de2552e61f5dc3b776caa60d2e3137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:55:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30c77b22e67a4f1fa905a0be01aa21f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89668434db7a41cca595a50badcf0115 : SUCCESS in 6m 21s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a1dc66e00f6848f7afcbf38f67c98fac : SUCCESS in 8m 02s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a7556f6df8044e39800d57bc40ad4f16 : SUCCESS in 3m 58s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bb24fcc6b12404183ec0fed2dcad617 : SUCCESS in 2m 39s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/a2cbff98d8394de29315bda59b16ae4c : SUCCESS in 3m 16s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/2c5ebe5a95fd459aba92b8f6c863032b : SUCCESS in 3m 02s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/69e763e077b944459eafd8c2b06ce88c : SUCCESS in 2m 53s\n- rally-tox-py311-sqlalchemy14 https://zuul.opendev.org/t/openstack/build/2774cc1298834ec78151aca97cfab8f4 : SUCCESS in 3m 13s\n- rally-tox-py311-sqlalchemy2 https://zuul.opendev.org/t/openstack/build/70108f78d43d472b9bedfe62b8f2b183 : SUCCESS in 2m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/46cf3d2de19c4ed5af6876f1e3f95734 : SUCCESS in 5m 47s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d2b0c081dd4446ebb4f6fd748faf6d62 : SUCCESS in 5m 57s\n- rally-install-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f7d8c4881b24318a87b2efdc1006de3 : SUCCESS in 5m 38s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5090e3baff241aea7dadb025726ac3d : SUCCESS in 5m 43s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/806f1f21459d4b77b0942126eb70b480 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a536fe9b84a8e68c8950797c268708b2c7265c16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:55:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06cd8ce716896813f16ebeac921263900497ffc1","revisions":{"c82e15099e397cdadd8487d354a493a16f3ffe80":{"kind":"REWORK","_number":1,"created":"2024-07-23 14:54:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/41/924741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/41/924741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/924741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/924741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/924741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/41/924741/1"}}},"commit":{"parents":[{"commit":"4e7a4e54142e134fc6e88a66014a1f36005d3c79","subject":"Merge \"Replace retired oslosphinx\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e7a4e54142e134fc6e88a66014a1f36005d3c79"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 14:53:08.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 14:53:59.000000000","tz":120},"subject":"Python 3.14: do not use deprecated ast.NameConstant","message":"Python 3.14: do not use deprecated ast.NameConstant\n\nast.NameConstant has been deprecated since Python 3.8, and will be\nremoved in Python 3.14.\n\nChange-Id: I5909ad0222e1ce388ec7593fadbb04cb5e4a0ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c82e15099e397cdadd8487d354a493a16f3ffe80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c82e15099e397cdadd8487d354a493a16f3ffe80"}]},"branch":"refs/heads/master"},"06cd8ce716896813f16ebeac921263900497ffc1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-23 14:54:53.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/41/924741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/41/924741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/924741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/924741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/41/924741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/41/924741/2"}}},"commit":{"parents":[{"commit":"4e7a4e54142e134fc6e88a66014a1f36005d3c79","subject":"Merge \"Replace retired oslosphinx\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e7a4e54142e134fc6e88a66014a1f36005d3c79"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 14:53:08.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 14:54:39.000000000","tz":120},"subject":"Python 3.14: do not use deprecated ast.Str","message":"Python 3.14: do not use deprecated ast.Str\n\nast.Str has been deprecated since Python 3.8, and will be removed in\nPython 3.14.\n\nChange-Id: I5909ad0222e1ce388ec7593fadbb04cb5e4a0ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/06cd8ce716896813f16ebeac921263900497ffc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/06cd8ce716896813f16ebeac921263900497ffc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
