)]}'
{"id":"openstack%2Frally~935598","triplet_id":"openstack%2Frally~master~I4acffa13b10f2017d661dee40743d8b35b4e9631","project":"openstack/rally","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-22 22:57:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4acffa13b10f2017d661dee40743d8b35b4e9631","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"ABANDONED","created":"2024-11-19 03:14:44.000000000","updated":"2024-11-22 22:57:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b7d54876be2dab330a1b68c57684a77ffe3795fb","_number":935598,"virtual_id_number":935598,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-19 03:35:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 03:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc5412ebf51da26ef80dc1173f0d519f12b83e05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-19 03:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"301ee8d23160536367b063838f57b0698665974b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 03:35:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dada74a01294016aa9859c9fc15272e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6285c7b8de4f9ebad0d2e10531b8cd : SUCCESS in 4m 45s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/60aca9e6876c4e18b164b1d16049b923 : SUCCESS in 8m 02s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0da8084c582e49089c28eba3bfde10d7 : RETRY_LIMIT in 1m 45s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5180a9b61c2e491e8ecd88ee03f5f8c2 : RETRY_LIMIT in 1m 49s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/9ec08f3aa54a4b95b340bd2310f79497 : SUCCESS in 3m 42s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/197d416a093f4a7b982f355c9c1d7c6e : SUCCESS in 3m 17s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/e2bb787602bf4100941a819b6ca70e56 : RETRY_LIMIT in 1m 49s\n- rally-tox-py311-sqlalchemy14 https://zuul.opendev.org/t/openstack/build/09931c0117b6447992a90b413c7359d7 : RETRY_LIMIT in 1m 56s\n- rally-tox-py311-sqlalchemy2 https://zuul.opendev.org/t/openstack/build/412d355526964075aa8dbbca98590a8c : RETRY_LIMIT in 1m 46s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/1df2cf905f694eff874adf33a946ea12 : RETRY_LIMIT in 1m 51s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/605c683e2f834a929e91bdf68c16ecd2 : RETRY_LIMIT in 1m 46s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/de4d472f16544a059a47ed46edf1ee25 : RETRY_LIMIT in 1m 55s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a49ce776ea4f969b69f696a0f92bce : SUCCESS in 6m 50s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0988cfa11a3d4b70ac88e61f4cbdf066 : RETRY_LIMIT in 1m 46s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/8797b96e1cc04c62b590d2fe45e3b68e : SUCCESS in 7m 36s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/918bfbf3cecb4f9d828b58f55e554ee1 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b7d54876be2dab330a1b68c57684a77ffe3795fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-22 22:57:15.000000000","message":"Abandoned\n\ndone in https://review.opendev.org/c/openstack/rally/+/917487","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78fc69e024f1ca88b4fcdccd13236eab64bb581c","revisions":{"78fc69e024f1ca88b4fcdccd13236eab64bb581c":{"kind":"REWORK","_number":1,"created":"2024-11-19 03:14:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/98/935598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/98/935598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/98/935598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/98/935598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/98/935598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/98/935598/1"}}},"commit":{"parents":[{"commit":"4e7a4e54142e134fc6e88a66014a1f36005d3c79","subject":"Merge \"Replace retired oslosphinx\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e7a4e54142e134fc6e88a66014a1f36005d3c79"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:14:10.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-19 03:14:43.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/935595\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I4acffa13b10f2017d661dee40743d8b35b4e9631\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/78fc69e024f1ca88b4fcdccd13236eab64bb581c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/78fc69e024f1ca88b4fcdccd13236eab64bb581c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
