)]}'
{"id":"openstack%2Frally~936935","triplet_id":"openstack%2Frally~master~I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5","project":"openstack/rally","branch":"master","topic":"no-phg_resources","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2025-07-28 09:50:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5","subject":"Use setuptools_scm instead of pbr and implement pep-517","status":"MERGED","created":"2024-12-03 11:43:36.000000000","updated":"2025-07-28 10:41:32.000000000","submitted":"2025-07-28 10:41:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936935-no-phg_resources","meta_rev_id":"6ca4d69114883c40d940d255c66cadaf2e0511f4","_number":936935,"virtual_id_number":936935,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-28 10:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-07-28 10:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-07-28 10:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-03 11:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc16042220c723be1605831307855b6c4097d84d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-12-03 11:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a2073ae8f48967a3ceaef39790e489c6d2d66ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 11:55:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b196d134e0b489e971fd5c9310ac49a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a1b1e5df54748fcab478899d840e734 : FAILURE in 1m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81da1da473d489fa91241ad6745e5b3 : FAILURE in 4m 02s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8c197b6094cd4426b6d38a52205010d8 : FAILURE in 3m 58s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8f098719d04c4dd5b3326d5f36fa49a4 : FAILURE in 3m 26s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed9efa8f534405fba7d5b955d49d78e : FAILURE in 3m 58s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/bb3ab69f7ff448cfa55005d6d3929c6c : FAILURE in 4m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/cd3f5c9a08f94ab09412e7ee48f4c8fe : FAILURE in 3m 45s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/359ac335de5d46d59cc39ba610beb6b5 : FAILURE in 2m 36s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/78e3d57799d64f76acace3f3eac8a065 : FAILURE in 3m 58s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e2a3e49c754f4f65abf959083f1cd585 : FAILURE in 3m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/26ed441bfca04ee293e77698026eb9e3 : FAILURE in 4m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a927baccd2da4809b53a477d3f5099f6 : FAILURE in 2m 44s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a9adf73056431193a7346ce9127fcc : FAILURE in 3m 01s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da39b0bf8df64bd98932422a0c38ae73 : FAILURE in 2m 29s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/0044307373dc4d009c1584ba14e1fb3a : FAILURE in 3m 27s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2b12e3c42a0243a89c33ae8d475a82db : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ccbe58847d5a4027de49302ae6d95cd1f9552a82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2024-12-03 12:41:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"522bd86e9df0a51bc9e32db0af62be29ae68a92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 12:50:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01551727d91840e18b79aa0af6177880\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a66f41becdf4a4c8006e5dc969282a9 : FAILURE in 1m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1011bd746e44425693d2599a06c8c01d : FAILURE in 4m 25s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/42dec46ab81c4188af7e713959041688 : FAILURE in 4m 24s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b3b8cc40a0e44a1c9b9467ad12e7d5b9 : FAILURE in 4m 27s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac986b8332f415c952d62ab743f4221 : SUCCESS in 3m 54s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/cbbf73ba5f454ab5a17d984df3f19d0e : FAILURE in 3m 00s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/15351f10aa2f4dd3a2cb706978687d53 : FAILURE in 2m 59s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/f960fa57e4cf49ae8a07833077d53565 : FAILURE in 3m 04s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/824fd12f854244cfa4826e58d4bd2268 : FAILURE in 3m 20s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/2218dc03d64c47c9bc5f2779095574e6 : FAILURE in 3m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9c47df888b82472ab266f4d28fdd7498 : FAILURE in 3m 44s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/511e866068cf46348cbd9f79f369f930 : FAILURE in 3m 26s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/269521e571eb49b5ab7cbd944652b7b8 : FAILURE in 2m 16s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7afaa753a29a48909075c6680a43c91f : FAILURE in 2m 37s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/4510854458f840b68e0e5a2ff18cfe7f : FAILURE in 3m 07s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6b2a71b9f6cd440fa1ce527d7e9e1a85 : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0e9757e63c5e4eee8041fa0877da8d178b5330c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-27 10:02:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"56f994a13f98452a7ddc0d3e808d12309fa441be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-27 10:09:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"03958a9cb2e918ea19c1042c20ba04a84a64a042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-27 10:28:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdb96b6edd174b4e8454079ec89674bc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e542178fdf9441a7b4841818884d40f9 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7b24b64c0741d3a094ab29f269a903 : FAILURE in 1m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e3d30e87e59343789e6b16f3d7a6cb8e : FAILURE in 4m 02s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/31bc78d4e21f4bad83076da61f1ebceb : FAILURE in 2m 55s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a96561f2c04ccf8cb65cc1c637bd63 : SUCCESS in 3m 37s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/981e88d6a0c6432bbbba0ae5d0ce009a : FAILURE in 2m 56s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a4264fb21a4b4cc08f1fb254f8c7d2b8 : FAILURE in 3m 03s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/9347b8a7945e4fffba074c1ead103784 : FAILURE in 2m 26s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/cf59b5e13c1846b1baf498ca861130d8 : FAILURE in 3m 56s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6ad975bff4894ef5b7fe136459461014 : SUCCESS in 5m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8937122ef2a64569a12f43f5e03ca72d : FAILURE in 2m 21s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0b20032d830b469dbf4715ecbf16741f : SUCCESS in 5m 57s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/862f0861965141f883067a7ed285def6 : SUCCESS in 8m 39s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22cc8369f2d44967afaa3e967fe49d24 : SUCCESS in 4m 55s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/206285f1eee140beacd73e36e8dda969 : SUCCESS in 5m 37s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/08c9455fdabd44448367b07d79d95ee1 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":4},{"id":"27584dfb70776467bc64121a83c0550f4ad2c817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-27 12:20:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"642a3793ad4b9e4f8786460c1d3cfee703543025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-27 12:35:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f43141d6554415ca849ee8429e390c7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5dba19537e344cbfbd6bf7292a6ece90 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f3754a96ea4355b819f3acf6787c7e : FAILURE in 4m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8ecda7ab48c341c6a4ff62c9c0d5addf : FAILURE in 2m 42s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8ff416810ed9479bb1025b6db0733851 : FAILURE in 1m 47s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b49e70fd486430682df4f1b49b319a5 : SUCCESS in 2m 17s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/2667e358b0164610929334b4fa08785e : SUCCESS in 5m 21s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/c7479627ecb64223903881503a2a9994 : SUCCESS in 1m 37s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e719819d8277448d8e1e7378132f41b0 : SUCCESS in 2m 30s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/96df549b32644f478f7d50a247bdc792 : SUCCESS in 1m 39s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/949bce30d52048a4adc97a0de3cdc405 : SUCCESS in 2m 15s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/50ee3004893240d08aa416ec9556c823 : FAILURE in 4m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a8a2fc76f0ac4feeabf4e8b785cd98a5 : SUCCESS in 5m 47s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88fb119adc8340199642faaa4647823b : SUCCESS in 6m 50s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a9f1f8ef92a4ef0ae5a784a0ccd859c : SUCCESS in 5m 38s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/e1a0bd8814fa4e83ba616af1eed18006 : SUCCESS in 7m 43s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/df2bf6ae800d48d18cefeb68ae34c8e6 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":5},{"id":"916a7519ede55d508f13356eda447b3a82e31132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-27 15:45:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"24eced3803ecb4e058233ceb861a959892acd12d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-27 15:46:12.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c907e29e33135255c9e31c25671ca1fba45ee7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-27 15:55:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdfeb219aff847729db65a0c408af1a9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9f9a33473c74bfe8d818faa4de04f7a : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc89b063c0084026acf859a9acdf2de9 : SUCCESS in 3m 27s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3170031d3e184cd893303d57b3439512 : FAILURE in 5m 10s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b91393cc991745d2ae4188e81864f8b6 : SUCCESS in 2m 47s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a19f22c51e475188645f2b5d8f606c : SUCCESS in 3m 32s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/899dda4b5c7f479c829c0f9e6c985694 : FAILURE in 1m 41s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/d8079f44fbfd4bc8a69a660f299e63ca : FAILURE in 2m 29s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/81378ddd03914fabb8a913b0562f0520 : FAILURE in 3m 48s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/9e7231c8161048bb950f7a2f50e1746e : FAILURE in 1m 42s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/d3a95e88ced448a0a0ef5a6cee6137ac : SUCCESS in 3m 26s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/cb1effc47abc4e7db0945c6319a04ecd : SUCCESS in 7m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a9182c68811948a0a353057a6a4dd94d : SUCCESS in 7m 12s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0662e1241b534200b6faff3d90abcdaf : SUCCESS in 7m 02s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2bb02d61c3e4d5b8db191c3f4a22ba6 : SUCCESS in 4m 53s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/c00a9d2a97894831ad9c91f5bb90f620 : SUCCESS in 5m 53s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/1523fadb0d564bb6905b1d0a15dc6910 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":6},{"id":"1d5ab236a7f24b4cfa8c8112c4476b35ff654420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 08:43:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"65695cb7cad9e08ae5b6b985c0b38b0a53e38a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 09:00:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/518cef19564d423dbc007d818b77a486\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e53da503a6b401faa02ac217bbdf8e1 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad046aa767d4fd18658edf88bf92e70 : SUCCESS in 3m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b13200ec13cb4f8ba03c103e1ef6a1e5 : FAILURE in 5m 10s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/6946e8d3d6094d46a85cbb1e3b7cd263 : SUCCESS in 2m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f69fb9dd37eb4d3dabe7b359097e98ae : SUCCESS in 3m 48s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/5d4700c3291a4c8485913b47235a785c : SUCCESS in 4m 10s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a53f5383dbf6409bb9a0c227500b80a1 : SUCCESS in 2m 32s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/b6194a3a0e2541eea8228073de280ffe : SUCCESS in 2m 39s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/c50a7d6c699c4899adc423c61dcad2eb : SUCCESS in 2m 38s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b4e52a601448454c9d05937a1d4ecd8f : SUCCESS in 2m 18s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c1458719ae0e4accaddb927cd0f68617 : SUCCESS in 6m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6979c80b01ee4c12bff61999ebd109fe : SUCCESS in 7m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/919f5709b41f46fab2968100d028e6ec : SUCCESS in 5m 39s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0feb0f1da1ba4d44a429ee74644b8a2e : SUCCESS in 5m 39s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/ee9567074e654aaaa2a813088576ad01 : SUCCESS in 7m 22s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/532c7ea1e9af42068743967e8211f788 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":7},{"id":"2eb095ee583316eac4cd7ab3277b6eb5a50608fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 09:23:53.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1865eebcee95fee2b84f13d2428c257140fc7fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 09:34:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8820be225436a1d8a0f8a2487e56f9d90414dd92","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 09:50:20.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4e7c26096261a3eea6bc4185ee48fffbf0235c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 09:51:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4472cf6cb6543229b70eac3b2b1c8f3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4eb7909a11e74442b8cc874632c03a62 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20945718d82d43f89e1ac271e8ee83e9 : SUCCESS in 4m 43s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/529d2e54565d4636b61b17c104564a4d : SUCCESS in 4m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2271038a41ba46f4a7c35d3700fa11b9 : SUCCESS in 5m 04s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/df0d33f08b2b4daf95a3a82a649fd5aa : SUCCESS in 3m 45s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/c268da6355f24025a05023a64600eb15 : SUCCESS in 5m 38s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/c6a98db361cc486186e01703a3dcae44 : SUCCESS in 2m 26s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/7a0679f7fbd64b78813011b82a282df2 : SUCCESS in 4m 06s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/1aa543ff2dd545a881db9849934a20a2 : SUCCESS in 2m 47s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/bb4151e859f845f1a4f236291f2919c6 : SUCCESS in 3m 55s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/2829fd55466747dfaee5636e3b440837 : SUCCESS in 8m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/53669e8d405e404eb5f4100037b3530c : SUCCESS in 6m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e3a7d3c0a08404dbfbfd6dab339ca96 : SUCCESS in 6m 13s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42c1cbbf46154d50a8024f6d215265cd : SUCCESS in 5m 47s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/208602fdbfb0499ebbccf04d34c0920b : SUCCESS in 6m 10s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/14df5b30ca784a858d8564ac00b1f38c : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":9},{"id":"7c00468fc248da3af33de7055b13e29c92f1cfb6","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 10:23:34.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8354dd09ade044ede4f56d95aef10374a254b8d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 10:23:44.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"727a5e0552ac083e9230fa5abc641bbb4404b7d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 10:41:32.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abcd2771472e4d00aa692764c0fb65ec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2eb5c0c6cd454256a3016934ddcccb25 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591dbd06b66448ecb50aeed0e508d050 : SUCCESS in 5m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/5f4c74da799f4ac6b5eb435570157666 : SUCCESS in 4m 44s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/fb36dba4d304415ab0075d564464ae4d : SUCCESS in 3m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee94fe7b753544de81f1782dce8aabb5 : SUCCESS in 4m 09s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/1e3327929ab9448c90b539f4e6eec668 : SUCCESS in 2m 31s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/2eb73b57d4b542159d14ee45ff71a4a4 : SUCCESS in 2m 41s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/28ea703ac5444afda660c22725e77e5b : SUCCESS in 2m 29s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/41f1ca2a576d406ea3d5e97fe620ab26 : SUCCESS in 2m 48s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/1f28ba9407ed46e6b9f6050219399d99 : SUCCESS in 2m 53s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/48c8f530f5d640759f91ad67f75d1cdd : SUCCESS in 8m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6cb5b5ae4ac941068860f2d288b3dd33 : SUCCESS in 7m 23s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0651713d873f46558f9494c42b6f1597 : SUCCESS in 5m 18s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca134ec61932422ea18ae7830b5fe365 : SUCCESS in 5m 41s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/b3d662c367cd482fa4c1d3731f9c807e : SUCCESS in 8m 36s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/b124892b2c3f41ce92b935e5980cce94 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":9},{"id":"6ca4d69114883c40d940d255c66cadaf2e0511f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 10:41:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"62eef988505bbbd92d6415d521f88caf30d2125b","revisions":{"cc1a3c3796efef9caf78da513d8216c4aa723656":{"kind":"REWORK","_number":1,"created":"2024-12-03 11:43:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/1"}}},"commit":{"parents":[{"commit":"087676be40e300d2d63424f9cbd6422198191a65","subject":"Update python versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/087676be40e300d2d63424f9cbd6422198191a65"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:42:54.000000000","tz":60},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cc1a3c3796efef9caf78da513d8216c4aa723656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cc1a3c3796efef9caf78da513d8216c4aa723656"}]},"branch":"refs/heads/master"},"b0efa69e5637c222dead26bea17fcc046bb9f0bb":{"kind":"REWORK","_number":2,"created":"2024-12-03 12:41:26.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/2"}}},"commit":{"parents":[{"commit":"087676be40e300d2d63424f9cbd6422198191a65","subject":"Update python versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/087676be40e300d2d63424f9cbd6422198191a65"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 12:19:48.000000000","tz":60},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b0efa69e5637c222dead26bea17fcc046bb9f0bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b0efa69e5637c222dead26bea17fcc046bb9f0bb"}]},"branch":"refs/heads/master"},"13a143c19ba1368d70dbced3df0c3a1f513a49ea":{"kind":"REWORK","_number":3,"created":"2025-07-27 10:02:53.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/3"}}},"commit":{"parents":[{"commit":"dcba35a3681c9a6254685790aa672d3db338d975","subject":"Replace deprecated abc.abstractproperty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/dcba35a3681c9a6254685790aa672d3db338d975"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 10:02:35.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/13a143c19ba1368d70dbced3df0c3a1f513a49ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/13a143c19ba1368d70dbced3df0c3a1f513a49ea"}]},"branch":"refs/heads/master"},"2706ace72008e7b6bdd8bf89f2d2b02786630234":{"kind":"REWORK","_number":4,"created":"2025-07-27 10:09:09.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/4"}}},"commit":{"parents":[{"commit":"dcba35a3681c9a6254685790aa672d3db338d975","subject":"Replace deprecated abc.abstractproperty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/dcba35a3681c9a6254685790aa672d3db338d975"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 10:09:02.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2706ace72008e7b6bdd8bf89f2d2b02786630234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2706ace72008e7b6bdd8bf89f2d2b02786630234"}]},"branch":"refs/heads/master"},"9165767f1fa7974aea2b479a9cb13dcf37006470":{"kind":"REWORK","_number":5,"created":"2025-07-27 12:20:48.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/5"}}},"commit":{"parents":[{"commit":"8f950742bcb6096d8ed316f109730d777b8926a4","subject":"Remove usage of deprecated pkg_resources lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8f950742bcb6096d8ed316f109730d777b8926a4"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 12:20:36.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9165767f1fa7974aea2b479a9cb13dcf37006470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9165767f1fa7974aea2b479a9cb13dcf37006470"}]},"branch":"refs/heads/master"},"7f2cdca6d665c7677ac5806163690d0ee49681c8":{"kind":"REWORK","_number":6,"created":"2025-07-27 15:45:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/6"}}},"commit":{"parents":[{"commit":"5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521","subject":"Remove usage of deprecated pkg_resources lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 15:44:09.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7f2cdca6d665c7677ac5806163690d0ee49681c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7f2cdca6d665c7677ac5806163690d0ee49681c8"}]},"branch":"refs/heads/master"},"7e5d8aa7ff882b79287fa152e8d05b4b4e4ed927":{"kind":"REWORK","_number":7,"created":"2025-07-28 08:43:10.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/7"}}},"commit":{"parents":[{"commit":"5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521","subject":"Remove usage of deprecated pkg_resources lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 08:42:46.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5d8aa7ff882b79287fa152e8d05b4b4e4ed927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7e5d8aa7ff882b79287fa152e8d05b4b4e4ed927"}]},"branch":"refs/heads/master"},"7156276751924069b944792c01c66ee8802d3531":{"kind":"REWORK","_number":8,"created":"2025-07-28 09:23:53.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/8"}}},"commit":{"parents":[{"commit":"5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521","subject":"Remove usage of deprecated pkg_resources lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 09:22:12.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7156276751924069b944792c01c66ee8802d3531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7156276751924069b944792c01c66ee8802d3531"}]},"branch":"refs/heads/master"},"62eef988505bbbd92d6415d521f88caf30d2125b":{"kind":"REWORK","_number":9,"created":"2025-07-28 09:34:50.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/35/936935/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/35/936935/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/35/936935/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/35/936935/9"}}},"commit":{"parents":[{"commit":"5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521","subject":"Remove usage of deprecated pkg_resources lib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5e83ecf74d4a4aa6bde9460ac55d6ac8862ec521"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2024-12-03 11:41:00.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 09:34:00.000000000","tz":120},"subject":"Use setuptools_scm instead of pbr and implement pep-517","message":"Use setuptools_scm instead of pbr and implement pep-517\n\nThis removes setup.cfg and setup.py files by replacing them by\na single pyproject.toml\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I220dafc2cfb0bbd92249efdf2ad713afe1e6d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/62eef988505bbbd92d6415d521f88caf30d2125b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/62eef988505bbbd92d6415d521f88caf30d2125b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
