)]}'
{"id":"openstack%2Frally~943096","triplet_id":"openstack%2Frally~master~I7609435c0d50bd60d825d3b6372810372ecc1a19","project":"openstack/rally","branch":"master","topic":"ignore-list","hashtags":[],"change_id":"I7609435c0d50bd60d825d3b6372810372ecc1a19","subject":"[ci] Extend deprecateion ignore list with eventuletutils module","status":"MERGED","created":"2025-03-01 12:39:21.000000000","updated":"2025-03-01 13:33:35.000000000","submitted":"2025-03-01 13:33:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943096-ignore-list","meta_rev_id":"17d7906f496c5e1cb9b8a0d1fc6633b31f389bc6","_number":943096,"virtual_id_number":943096,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-01 13:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-03-01 13:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-03-01 13:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-01 13:18:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bffc6687b9301cead68f679ca76d3763e861c08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-03-01 12:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efa5489e991697460ecfcc791a2f7de5d9c54145","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-03-01 13:04:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f035e42cbfc360562e954a371b2f1d7449b97b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 13:18:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bbcb5d67cb1493e984af2faee9aa222\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf66f6cf3164001aabbfa1811583cad : SUCCESS in 4m 47s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3313da15b3fb42289f289ba8d15f704d : SUCCESS in 5m 01s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/bef1a28ae5bf47c19fcdd9ef29316c9c : SUCCESS in 3m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae115f3d55cf4aa195907920a895028d : SUCCESS in 2m 58s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/be5e9e928d1447f49dc656689f6bb3fc : SUCCESS in 2m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/c0e17256420840f1bdf311941bfe0edb : SUCCESS in 2m 43s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/eee588db8f574b329d5657679ab4b01e : SUCCESS in 5m 04s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/f814d8e49eb7425a9c428825ce96395f : SUCCESS in 3m 27s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/75a7ea59ffe34f8c8a14a3a5eba73a71 : SUCCESS in 2m 43s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8bd43664d6454043959def704818ea49 : SUCCESS in 6m 02s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/df5c6d1fc92b4a7aa0af0be6bdb3f97c : SUCCESS in 6m 12s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da2c82eacd194b1ab96beec441b23a83 : SUCCESS in 8m 36s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f72d28ed7db4ccd9781d709b5edd487 : SUCCESS in 7m 20s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/4a7ab7eb700b46dba0b1c7f5d103dfcd : SUCCESS in 6m 07s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6a942924543f4cc4a1e2e40adf9a9b26 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2a691daa1bea8a263b4934f92b5d24fe0ae7bcc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 13:18:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f52f0c9f9fd8604dcb744238cb29c12ea7c7281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 13:33:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/682c93af155d40699686122ad273b009\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913d8992cf2444918973b8eaf84ac1d6 : SUCCESS in 2m 36s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a58119ba2707418da36b53f5e50234c7 : SUCCESS in 3m 55s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ce36d22290054f788622bed7de7844b0 : SUCCESS in 2m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea6f4bca3974fb58deeb1b9d76f0e13 : SUCCESS in 1m 56s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/6e1b606f95304893b89797a956c216bf : SUCCESS in 2m 54s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/369281203b52437aaef70f34210f4023 : SUCCESS in 2m 25s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/d1d8adc243464403b9e2c614681c7771 : SUCCESS in 2m 32s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/ab9e3b9190974e5abd7ce6254241138c : SUCCESS in 2m 52s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/0b94b594814345ecbfb95191447ed3ab : SUCCESS in 2m 33s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c9c0aa5ad2174efd81f46db55718dcad : SUCCESS in 8m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/844def5ca3c047a7bcb868b972d3ce52 : SUCCESS in 6m 17s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc839b5ed8e14507ae28f21ee137a692 : SUCCESS in 7m 06s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31d64853aa814650863b1f990ef1e9b0 : SUCCESS in 6m 09s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/96a64db226fc4363a30e3dfb04fa5ab6 : SUCCESS in 6m 47s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/56c0114c0af94dbe89a337cc6770cb36 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"17d7906f496c5e1cb9b8a0d1fc6633b31f389bc6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 13:33:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4917ee6ea4a40a273967e97a7069fe9c45e14834","revisions":{"4917ee6ea4a40a273967e97a7069fe9c45e14834":{"kind":"REWORK","_number":1,"created":"2025-03-01 12:39:21.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/96/943096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/96/943096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/943096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/943096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/96/943096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/96/943096/1"}}},"commit":{"parents":[{"commit":"087676be40e300d2d63424f9cbd6422198191a65","subject":"Update python versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/087676be40e300d2d63424f9cbd6422198191a65"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-03-01 12:38:47.000000000","tz":60},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-03-01 12:38:47.000000000","tz":60},"subject":"[ci] Extend deprecateion ignore list with eventuletutils module","message":"[ci] Extend deprecateion ignore list with eventuletutils module\n\nWe do not use this module\n\nChange-Id: I7609435c0d50bd60d825d3b6372810372ecc1a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4917ee6ea4a40a273967e97a7069fe9c45e14834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4917ee6ea4a40a273967e97a7069fe9c45e14834"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
