)]}'
{"id":"openstack%2Frally~943112","triplet_id":"openstack%2Frally~master~Ie0e485fb753f271d8dd9edd93137700b5d9b0615","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-02 08:40:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0e485fb753f271d8dd9edd93137700b5d9b0615","subject":"Replace deprecated abc.abstractproperty","status":"MERGED","created":"2025-03-02 04:00:19.000000000","updated":"2025-03-02 08:40:23.000000000","submitted":"2025-03-02 08:40:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943112","meta_rev_id":"dfe49081c9cea9293e1616bfe3160feb7bc9ac56","_number":943112,"virtual_id_number":943112,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-02 08:40:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-03-02 08:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-03-02 08:20:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-02 04:32:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-02 08:20:47.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"62e5d86908fd79c061d5a1f7f79279aad759ed6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 04:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8da232e59f2f1f4257b9629f5be9f8aff5d9005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 04:05:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45fcf44bc556557023f756dadfdcd4f32060b398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 04:32:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6144edce639f47feb71ccab8e93f9744\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db1f1d8232543d1a85b7a0048381b22 : SUCCESS in 5m 02s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/131d75e686554193873562cab3fe5a5c : SUCCESS in 4m 17s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ec3e78fcc7574809b8001c3e3b935850 : SUCCESS in 5m 51s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7448389bb848189259f22a722de268 : SUCCESS in 2m 47s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/736a216a61434693b682079d4ae3fe2a : SUCCESS in 4m 15s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8ac08de35ba74b74844c0b4af86dc9cd : SUCCESS in 2m 37s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/6b8dd3f1fac949f29bb4235718415f5e : SUCCESS in 2m 48s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/cf027a665e194073a0d91b868e18f0ff : SUCCESS in 3m 01s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/99b2f290347a4f62adf73a3dbbc0f878 : SUCCESS in 2m 48s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9ce625562aa447de934280c033bb1006 : SUCCESS in 5m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/960a6fccc3994e69b11af14e4bd01c61 : SUCCESS in 6m 06s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c661ec5904424f7c9f1e3d7b96109f65 : SUCCESS in 5m 11s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f37f5af5579041bda6a8c1aa7a2aff8f : SUCCESS in 6m 03s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/e712220ceb584093869ad39d95e39231 : SUCCESS in 6m 01s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4973395f8eaf48aea18649aec8eff240 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"9ef06d138c7adfd5b67c6dbf165c4038f9f759ad","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-03-02 08:20:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4c92dac46d1e0db5b173ff6d7be9b11b6c67377a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 08:21:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"149a58fdd0fc21b0e0721f94f7584863145b80de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 08:40:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084930ed83d54235a9978f7c4598788f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25662f0be90c48d4a92c79f5085daca0 : SUCCESS in 3m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/376a074054784d61b58e987c3cddc4b0 : SUCCESS in 6m 18s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0931b432f3b844ec813d50d7f0a78ed1 : SUCCESS in 3m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/11cb6a8f6887443f830818ca6263ea07 : SUCCESS in 3m 43s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/6ac5b6693c0f4285a3096ffcc65770b1 : SUCCESS in 3m 43s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/1e4142f1bff245b084d12db770d2f755 : SUCCESS in 3m 42s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/ff671f22780a441191a7e8a6b8a4f957 : SUCCESS in 5m 03s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/489aaf3d8a344c5a9ae1ff2bdd314adb : SUCCESS in 3m 51s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/86aa32001d64477c803f4d06644767f2 : SUCCESS in 3m 39s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/5af9dc2459bf46a49a47e268a1a1a324 : SUCCESS in 7m 03s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/55d6d6c5082b4203ae0430b2d395c069 : SUCCESS in 5m 55s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/921fc676697741a88d894cf8b08a7c91 : SUCCESS in 5m 27s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b7e5f4dcd5d4084a501d25c41a7d0c9 : SUCCESS in 5m 49s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/fefccf9f16884ed7a35c6363286b2201 : SUCCESS in 5m 58s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/0fffe3ef567b45348f60b42189107505 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dfe49081c9cea9293e1616bfe3160feb7bc9ac56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 08:40:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dcba35a3681c9a6254685790aa672d3db338d975","revisions":{"3c2637eba031f50b3f0840286d78f71e9c12c2a0":{"kind":"REWORK","_number":1,"created":"2025-03-02 04:00:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/943112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/12/943112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/943112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/943112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/943112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/12/943112/1"}}},"commit":{"parents":[{"commit":"6172f9678a17ee055ce5e744883b4cd61b592906","subject":"Merge \"Get rid of pytest launcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6172f9678a17ee055ce5e744883b4cd61b592906"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 03:59:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 03:59:56.000000000","tz":540},"subject":"Replace deprecated abc.abstractproperty","message":"Replace deprecated abc.abstractproperty\n\nIt was deprecated in Python 3.3 [1].\n\n[1] https://docs.python.org/3.13/whatsnew/3.3.html#abc\n\nChange-Id: Ie0e485fb753f271d8dd9edd93137700b5d9b0615\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c2637eba031f50b3f0840286d78f71e9c12c2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c2637eba031f50b3f0840286d78f71e9c12c2a0"}]},"branch":"refs/heads/master"},"dcba35a3681c9a6254685790aa672d3db338d975":{"kind":"REWORK","_number":2,"created":"2025-03-02 04:05:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/943112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/12/943112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/943112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/943112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/943112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/12/943112/2"}}},"commit":{"parents":[{"commit":"6172f9678a17ee055ce5e744883b4cd61b592906","subject":"Merge \"Get rid of pytest launcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6172f9678a17ee055ce5e744883b4cd61b592906"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 03:59:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:05:21.000000000","tz":540},"subject":"Replace deprecated abc.abstractproperty","message":"Replace deprecated abc.abstractproperty\n\nIt was deprecated in Python 3.3 [1].\n\n[1] https://docs.python.org/3.13/whatsnew/3.3.html#abc\n\nChange-Id: Ie0e485fb753f271d8dd9edd93137700b5d9b0615\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/dcba35a3681c9a6254685790aa672d3db338d975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/dcba35a3681c9a6254685790aa672d3db338d975"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
