)]}'
{"id":"openstack%2Frally~956005","triplet_id":"openstack%2Frally~master~I6c2dcf5d89926a30e8a7d9a22acaa6c8687f0836","project":"openstack/rally","branch":"master","topic":"basic-mypy","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2025-07-28 14:11:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I6c2dcf5d89926a30e8a7d9a22acaa6c8687f0836","subject":"Start covering project with type hints","status":"MERGED","created":"2025-07-28 12:20:51.000000000","updated":"2025-07-28 14:22:59.000000000","submitted":"2025-07-28 14:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956005-basic-mypy","meta_rev_id":"17bd6d4ccd10d66420ea0851fd0cc07403120683","_number":956005,"virtual_id_number":956005,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-28 14:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-07-28 14:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-07-28 14:11:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-28 12:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51ba6bb041bcf2aa9d01e5da4d3b186deb8d574d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 12:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41dca0f179007b395d6906517635acf45ada2131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 12:46:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c07470171dab4fcf8d0ba2ead085b635\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0414093b2524968a3d9b402c1ff266c : SUCCESS in 2m 23s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b91c7685892343ff80a4b3071306945c : FAILURE in 6m 21s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ef72aecb0ce04e87ab1d73ed4d3311a5 : SUCCESS in 5m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8258d8bb9ae045ed92d5895f5406dc49 : FAILURE in 1m 38s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/5c66dd3437d54bb48ad1cd6a12711dbe : SUCCESS in 4m 16s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/6f5a9214f36a456bb75fa1b2bb8bf64a : SUCCESS in 2m 43s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/56f2d883cc334eb09d812af7b822eb2a : SUCCESS in 1m 39s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/19b57d2b4c834cb8a3a21d04993bcbe3 : SUCCESS in 1m 44s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/65f473ddbf914e7dad2ac1248e8f26fe : SUCCESS in 2m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ec36a1609d354cbeb8456c14464a5ac2 : SUCCESS in 6m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ab495149b3514bb2a005a5803087d420 : SUCCESS in 7m 24s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90d6c3bf461b4f7cafa994d61f8905d9 : SUCCESS in 5m 21s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff1375827a864b5ab633bd3a94077300 : SUCCESS in 6m 04s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/bb7190e9740045acbe1c5b886ea0b8f0 : SUCCESS in 6m 07s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6e8e0410a9fc4b078f5e3934cbf50be1 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"eb1c239dfb26f5d0d2ef859fc7b97f09e39b0ded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 13:18:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"47601858d12fd1e4913d58266dee3d077be01df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 13:32:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02a6215c433144d0be41a7f35b418f77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3818f2980ab642d09e0d78d74de77145 : SUCCESS in 6m 56s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/134c31d1c22b462691dfb8e2c04b6ffd : FAILURE in 6m 01s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a355c5d57ae54db780fcae6f58e57dbb : SUCCESS in 4m 49s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/51fd17ca0d4f46848c905c1fa9afed26 : SUCCESS in 3m 28s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/090c913cc11444f5832e4de7c0a86109 : SUCCESS in 3m 55s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/598e008378ba4eb09a8457d0ab1eb85e : SUCCESS in 2m 54s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8a41bee7388246a49091fda7699f7245 : SUCCESS in 2m 28s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/ad3c29d815c8479e853fe232cc79bbf1 : SUCCESS in 2m 51s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/74473723752b4915b88f4993845310a6 : SUCCESS in 2m 56s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/eeae2964e382454e9bef6b939f503ea0 : SUCCESS in 2m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f2c6eeab0ec1415a90cde28acf6c2552 : SUCCESS in 7m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/511a900c46c04836a92ec39fc67764fe : SUCCESS in 6m 08s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebf8e6e334a849aca3b26cb550a4ca19 : SUCCESS in 5m 27s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3f00a08a97e4d869642e11cce2d9712 : SUCCESS in 5m 49s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/1476f9c7757f4290865b6f78c2b27b50 : SUCCESS in 7m 44s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6366793407ca4d74be0f3b491ca7220b : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a011aecc49df8249bc91effbb79ca0f6218f77da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 13:33:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"67a07f477f2571f6165431c331b1e533dd9900fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 13:42:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30f4eb8b9ff4419a8d97ba9b486a9e6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277b3f2227f348feb787436594177a1f : SUCCESS in 4m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2b1a1e420d2348348aef4dd8140ae59e : FAILURE in 4m 42s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e0e7400ebf2c46f2b44c06ac99f55207 : SUCCESS in 3m 23s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/10e7ec9542f5462ba7cd9f91042d1ad7 : SUCCESS in 3m 16s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29fe24687b5417d82a37b0296aa19be : SUCCESS in 3m 04s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/c706e92c70bc45e6911d3c49c2e9eef8 : SUCCESS in 3m 05s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/3543472204114eeab58162b6ce8c3626 : SUCCESS in 3m 04s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/00a0a93e001e4019b55c0bd9f9859f21 : SUCCESS in 3m 27s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/f5edffa8ce404e35ac3dc1411158a42f : SUCCESS in 3m 14s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/028f237659b748c59526d32bb63427e3 : SUCCESS in 3m 05s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d059ba23c2d74d18b5de2be43311a566 : SUCCESS in 5m 53s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/85a17afdea2e40c58bbf058dfe913c96 : SUCCESS in 6m 30s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/974b6a24e6184801834128ff2a55544f : SUCCESS in 6m 09s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac7b60b05050473b96c2e285437b05a8 : SUCCESS in 6m 31s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/47817e870f1d4954bb9eb165460662d0 : SUCCESS in 7m 14s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e81d4b95057245d68f681ff2720eb009 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":3},{"id":"93da2354a734aad76eaed0fb5a3355d812647841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 13:57:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c55fe2c2bdb80c5b22830a43a68bfcee49789c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:08:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ca9f985494433dbdd9c415873dc59a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84fe8af957ef45f39d65c05d1a7bcfe0 : SUCCESS in 4m 11s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e0ff857c3eb54e2a86c806fa90ca3899 : SUCCESS in 2m 54s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/602061d806ae49b88f489e554cd2b6c2 : SUCCESS in 2m 28s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/902d4980eeb5461eb772591e6ecce8c1 : SUCCESS in 2m 03s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6cc00f78504607a4658791d52d860f : SUCCESS in 1m 44s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/2185d232f701482d83cd00801be346b6 : SUCCESS in 2m 35s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/5b1e7b434cce412f80dd3624f2853faa : SUCCESS in 1m 45s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/0a3c103dd6784132822e010898123cd9 : SUCCESS in 2m 19s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/3b755249756c4102839ac2cdcd2ba158 : SUCCESS in 1m 58s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/42a5bba0d5ea4c149024a7494df6f038 : SUCCESS in 1m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7d3d46e51762496a856e7ec63da7f4e7 : SUCCESS in 5m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f6bec0f970ab478799e45e447019d223 : SUCCESS in 5m 19s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2c2467421d1431ea9071e5bcd6eb08b : SUCCESS in 5m 25s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c283f1b5c43e481f978ec3640fc0dbee : SUCCESS in 5m 18s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/453f5676133847b7a9d4fbf789cd3b44 : SUCCESS in 6m 10s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/9106860ddcb5423aa56f801e29bb60c8 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"852963143444b5089e00302ef57fb96bb87a3dc9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-28 14:11:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b48f69dd1f0127280fc503b470153c123f9a783b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:12:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b3cae9a205d3084114e6c6701e2668e231c65fa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:22:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/799cbce3df5146c7a7b7213a50bd3dfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af3427cb4864666b8a9a34ddf9f9404 : SUCCESS in 4m 03s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b15c610e0d454cac8b611529bf23d6cb : SUCCESS in 4m 58s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/652a12612d4748fe9f4b4976d8f41239 : SUCCESS in 5m 05s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/75a27c0aabec46f29ff71f9547e7b492 : SUCCESS in 3m 14s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d74317816149b4ad6e7da423dc9653 : SUCCESS in 4m 04s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/a6aad6e4471541bc8252cae6868d80ec : SUCCESS in 2m 38s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/3f6d8d25304940fa96c788751903ac04 : SUCCESS in 4m 08s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/72fd9d35a7c246f3ae48ababa4292147 : SUCCESS in 4m 14s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/a07bb9eb609d447a85eb44de6e8edeb8 : SUCCESS in 2m 57s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/bfb5651d6494473a842bf6a4891c31c5 : SUCCESS in 2m 34s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9d18962f5e4f4549b3125f93eed111a1 : SUCCESS in 5m 50s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a2ae367247654cf9ace79f0bb366b97c : SUCCESS in 7m 22s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4beb44111bad4c0e9790e3acbdd94259 : SUCCESS in 5m 11s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0430c6a1633c4c9a9488253312d93131 : SUCCESS in 6m 40s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/36635b278a274d269cf2c8c07e0d75d0 : SUCCESS in 6m 08s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/eca2fe85fa32466fa53f28b61e2bef8a : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":4},{"id":"17bd6d4ccd10d66420ea0851fd0cc07403120683","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:22:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"765d903321e695e04b53e74f44a00e05d6cc4e73","revisions":{"f7198bb431c3e13ad1d3bddf7bc2c837cf31bace":{"kind":"REWORK","_number":1,"created":"2025-07-28 12:20:51.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/05/956005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/05/956005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/05/956005/1"}}},"commit":{"parents":[{"commit":"62eef988505bbbd92d6415d521f88caf30d2125b","subject":"Use setuptools_scm instead of pbr and implement pep-517","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/62eef988505bbbd92d6415d521f88caf30d2125b"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 10:44:23.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 12:20:43.000000000","tz":120},"subject":"Start covering project with type hints","message":"Start covering project with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I6c2dcf5d89926a30e8a7d9a22acaa6c8687f0836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f7198bb431c3e13ad1d3bddf7bc2c837cf31bace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f7198bb431c3e13ad1d3bddf7bc2c837cf31bace"}]},"branch":"refs/heads/master"},"f1fc92a0f57364c8e8171a557837c04a22850119":{"kind":"REWORK","_number":2,"created":"2025-07-28 13:18:34.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/05/956005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/05/956005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/05/956005/2"}}},"commit":{"parents":[{"commit":"62eef988505bbbd92d6415d521f88caf30d2125b","subject":"Use setuptools_scm instead of pbr and implement pep-517","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/62eef988505bbbd92d6415d521f88caf30d2125b"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 10:44:23.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 13:18:24.000000000","tz":120},"subject":"Start covering project with type hints","message":"Start covering project with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I6c2dcf5d89926a30e8a7d9a22acaa6c8687f0836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1fc92a0f57364c8e8171a557837c04a22850119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1fc92a0f57364c8e8171a557837c04a22850119"}]},"branch":"refs/heads/master"},"7b5eff7deb3fadc7f4f482e537c53a02546a6b86":{"kind":"REWORK","_number":3,"created":"2025-07-28 13:33:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/05/956005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/05/956005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/05/956005/3"}}},"commit":{"parents":[{"commit":"62eef988505bbbd92d6415d521f88caf30d2125b","subject":"Use setuptools_scm instead of pbr and implement pep-517","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/62eef988505bbbd92d6415d521f88caf30d2125b"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 10:44:23.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 13:33:38.000000000","tz":120},"subject":"Start covering project with type hints","message":"Start covering project with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I6c2dcf5d89926a30e8a7d9a22acaa6c8687f0836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7b5eff7deb3fadc7f4f482e537c53a02546a6b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/7b5eff7deb3fadc7f4f482e537c53a02546a6b86"}]},"branch":"refs/heads/master"},"765d903321e695e04b53e74f44a00e05d6cc4e73":{"kind":"REWORK","_number":4,"created":"2025-07-28 13:57:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/05/956005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/05/956005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/05/956005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/05/956005/4"}}},"commit":{"parents":[{"commit":"62eef988505bbbd92d6415d521f88caf30d2125b","subject":"Use setuptools_scm instead of pbr and implement pep-517","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/62eef988505bbbd92d6415d521f88caf30d2125b"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-27 10:44:23.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-28 13:54:21.000000000","tz":120},"subject":"Start covering project with type hints","message":"Start covering project with type hints\n\nPRAGMA: NO COVER\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: I6c2dcf5d89926a30e8a7d9a22acaa6c8687f0836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/765d903321e695e04b53e74f44a00e05d6cc4e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/765d903321e695e04b53e74f44a00e05d6cc4e73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
