)]}'
{"id":"openstack%2Frally~956132","triplet_id":"openstack%2Frally~master~Ib806217076ae85d1744301b381b17f1c7556e5ec","project":"openstack/rally","branch":"master","topic":"more-type-hints","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2025-07-30 11:27:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"Ib806217076ae85d1744301b381b17f1c7556e5ec","subject":"Cover more task related things with type hints","status":"MERGED","created":"2025-07-30 07:45:45.000000000","updated":"2025-07-30 11:57:51.000000000","submitted":"2025-07-30 11:57:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956132-more-type-hints","meta_rev_id":"eaa8d33e219d053b5d730e59fcd72b1b1b71f024","_number":956132,"virtual_id_number":956132,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-30 11:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-07-30 11:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-07-30 11:27:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 08:00:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f829ff7a1d13aeff76cedda5a2754e1bb53bce25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 07:45:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c46b346b5dc116fb3f5413f7aa953b6d17605ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 08:00:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2996d465eb9b467c841ed2df855a440c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b33b063d784b31b7c33a9a1948956c : SUCCESS in 4m 01s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ec24e18f3bd241dab1338efbc2fde237 : FAILURE in 4m 56s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/30e80a5293374a49ab285999d9c887cc : SUCCESS in 3m 29s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/dc12a726ed934919af05acb59461a210 : SUCCESS in 3m 17s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0df5e978d2444db9f01c390b3a3e69a : SUCCESS in 2m 42s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/98fdbaed4ab34d758ed724f42fcbba39 : FAILURE in 3m 27s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/30dc887317ef480484d8e6699cf0e31d : FAILURE in 3m 01s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/48310f88be1e4b2c931e2165b3dd7330 : FAILURE in 2m 53s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/7b41f09d9fc8481ca3db7013f91c8519 : FAILURE in 3m 16s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e486fd7ce7124040a66642e1b5d1460a : SUCCESS in 3m 59s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e9ac682797434e82b33d668803f1817c : SUCCESS in 5m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/20fa9ccabcd84dd7a490aacebc173362 : SUCCESS in 6m 13s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbb7782332164dc999b50d82004f41eb : SUCCESS in 5m 55s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc6bd4e0928641f29b43099d6800b7cd : SUCCESS in 7m 16s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/d4fb34919d384611ac4025f405bb2587 : SUCCESS in 6m 14s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/1bf4d4c913034f09bd1c4e103021223b : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2cda16abf4fbe69de97d916ace1abaf7bb02952d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 09:20:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4ab1494c485bf0176f480cbe5a698f0f6510654e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 09:47:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6245163ef28b47f48a5a74476a0a8974\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87273e5f14a4405eb2fd56a1988cd38d : SUCCESS in 5m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/354fd6a8581d4deeac2501f84536f9b2 : FAILURE in 2m 56s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8750faf6db134d67b2a7a4fb16ad1c5f : SUCCESS in 7m 04s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/5e7d9fead8f5419583b88f410dfdb73e : SUCCESS in 4m 12s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d165988a4f44774b5bc0211e98c0437 : SUCCESS in 2m 43s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/126b31da6d4f44d9b390020ac9a21e92 : FAILURE in 6m 00s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/ad9cfc35987a419594c36c66766e21a2 : FAILURE in 1m 41s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/f9ee7965cedc4b49b95a122e5c616920 : FAILURE in 4m 21s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/a513ecc3a1d3456fbd57b7d286bba0f7 : FAILURE in 3m 23s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/86f5080c70d64d85acc21b80b8b74327 : SUCCESS in 1m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/fb02d94b9e434219912aec3e18eaab71 : SUCCESS in 4m 28s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3bf20ef25c0849388558a655ab4b4fb3 : SUCCESS in 5m 04s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6113811489244b10bb59ad3221609d4f : SUCCESS in 4m 41s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b8c98f5f25417c9962c7b1bba0cb99 : SUCCESS in 4m 59s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/11af175d73b34a61ad252d2a417bcece : SUCCESS in 5m 09s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/f421b0c1be394b0da9c8543a515b1fc6 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0377a05480fb15d2231b4259a1d859f7b497525d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 10:00:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2573c700098780a252b05f194f3b113eb0d75924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 10:01:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d33f6220e9af5029e4e971cf9bc30ee7e10eee5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 10:13:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6852bb81bbe94767f6762e6a0c6a75ec254b903b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 10:39:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdd0ccce9cbb4170b9b81d29d035bfa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b68ca30e3b46c2bba74e1103d7d4e1 : SUCCESS in 4m 05s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/660318690bd64752bae483f4efaeea44 : SUCCESS in 4m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/872c6370e0f043e999f9fa0ec61dc6e5 : SUCCESS in 4m 03s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/196354d4dcc445dfba0b7f15ab7d4332 : SUCCESS in 3m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/733e9ec1c9c148959eae8eef8cf1e838 : FAILURE in 2m 18s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/817b05f2be584ea29e1a2960cf9f3935 : FAILURE in 1m 47s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/ba07d4e213234b698f31c09536be4f1f : FAILURE in 3m 26s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2af079af3f80412e80590e9117e6ed4e : FAILURE in 3m 16s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/06cd0c55fa35496c91a2d7b3c9239ab6 : FAILURE in 2m 29s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/1504c00585cc4adbad5b0f2b6ec8912e : SUCCESS in 2m 33s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/95ce5278b30a42708181b625bc29da81 : SUCCESS in 8m 07s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9228f1b0032b468fa2cfe2b23e297c86 : SUCCESS in 6m 05s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71b5b4914d5e47dba4668f552502a934 : SUCCESS in 5m 38s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d236fb39c3c4ec2916e109d447c93b4 : SUCCESS in 6m 41s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/3c12ec3344324e60b5aa2933d04ce618 : SUCCESS in 5m 13s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/653312cb592d4ea9ac9837aa403bab62 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"5dd962bd35e93a1329b18ea75095e2caa130e413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 10:52:07.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3b33faac9470db661e52424f803854835ef51aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:16:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd22132046084c76b35551a385cf448b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a26d905b7d4f23afae96a5d5dd4e61 : SUCCESS in 4m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/edb2a130992c4015a02bec0a174f226b : SUCCESS in 3m 40s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cef18e4bed7848f3a9c85e898bc73891 : SUCCESS in 3m 53s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/4622f4152dbc4ba8b9d5423d11061378 : SUCCESS in 3m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/833745c17c5045ee8e62b66132e938f7 : FAILURE in 2m 26s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/613305106cd44d3f84040b38b077af00 : SUCCESS in 2m 52s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/97639fa477b945d18eb409860e973c71 : SUCCESS in 2m 46s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/7552b87ac6f04d35856688c5ae8cce81 : SUCCESS in 1m 57s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/1bb250d8531e4cec8c00330a23fe72d7 : SUCCESS in 2m 54s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/857cfdb86fbd42eab870fdf626ab494f : SUCCESS in 1m 39s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8caf39d515d943068ee4997b158f8176 : SUCCESS in 8m 39s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2b2a5b15ffd94c4c9127b2ff3d0ed032 : SUCCESS in 6m 14s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e6f40c05fac484dbf7df0cea6bf12c2 : SUCCESS in 4m 47s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2778b8725d5a4812b3e91b630c427b0f : SUCCESS in 6m 24s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/522272f9482a4534b7f6b678dff95ecd : SUCCESS in 6m 36s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e498c98d72f7481b87687d7b3ead3da4 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":5},{"id":"029d1b6e5bb19e8e2475f75d6ab4f976c402c43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 11:26:25.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"fe8953a991a0246637ff6433e4ebded6d01f85fa","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 11:27:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"63980859d32958d02ca200d4b7df38c928edbcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:45:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e21322068f44289b9c80605cc5b73c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba36bad45fa4c6fac045687c2b39fd0 : SUCCESS in 4m 02s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/fa32015dcef04f28902dff742fa18ce6 : SUCCESS in 4m 32s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/90595e42a7f6484392f95a5045df78ea : SUCCESS in 3m 21s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/6cb3aaf0506e465ea470b28c101b2835 : SUCCESS in 3m 06s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3b3c3c47574ccda56891cda943fd67 : SUCCESS in 3m 26s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/fa54323b396241f8b8d07915eab0a06a : SUCCESS in 2m 38s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/6898e05aebd440aeb137316fe7c3183a : SUCCESS in 2m 30s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/da4a414d3b134405afc24770b03dea76 : SUCCESS in 2m 40s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/f7747f1ef2ae4fc798fa3242a628ecf5 : SUCCESS in 3m 56s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/56c522ef8e8942c3a778d6ab7aad05db : SUCCESS in 4m 04s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/002e2c1d2a7c41e081df7ea678441314 : SUCCESS in 6m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2c2849fd6cc44f47a6318d65a970caad : SUCCESS in 6m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/851e28f94bd347d78bd258e3c8fadc6e : SUCCESS in 5m 33s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75b1506967de439588135d6047fb41d1 : SUCCESS in 5m 59s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/20861774df1440cda352a74b939ccabd : SUCCESS in 7m 51s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/eb213cdeb74b4a6fa00ca09d4bba0c53 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":6},{"id":"4cb32ee6afb6c45fe9d71cb9704499bfd410223e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:45:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"819f63c19d2a20cc4e55a4387fd0c46652da43c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:57:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dfc8f10bfb74d6cbeeae56ced4ad8d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1379d75291c64d5eb8aa89a81beffffe : SUCCESS in 5m 14s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9f227e0115c742b18413a73e5c75b99e : SUCCESS in 4m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a5b01315bb85485a92c2f9906f80b4d1 : SUCCESS in 4m 55s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/e294b31cccfa46aeb84c275ed39f92cc : SUCCESS in 3m 44s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2618d80e58ea45b9ba9be96ad65f7627 : SUCCESS in 2m 59s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/0571091a712c4671b551a24519689b1b : SUCCESS in 4m 07s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9321f67230a948049b6597a2109472ca : SUCCESS in 3m 26s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/d92a83f12e004da8a6984c71a60f823f : SUCCESS in 3m 29s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/f982763d05b34c28974de2a7ade20b53 : SUCCESS in 3m 23s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e92fd59fbcdc41ce8f2fa599210f030e : SUCCESS in 3m 09s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/8aefea30c33c4c839b39f2ef224a2249 : SUCCESS in 7m 14s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/03e5130378a74a0e93d093d030705ac6 : SUCCESS in 6m 53s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/947f3c12b8d94a769f7717890e73b502 : SUCCESS in 6m 12s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc648a1152774c37969236ca00c9f150 : SUCCESS in 7m 22s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/2745cb9c4a254205ac722d766c2e2333 : SUCCESS in 6m 38s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4f6a02829b00446aa9b1ed9b864a1560 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":6},{"id":"eaa8d33e219d053b5d730e59fcd72b1b1b71f024","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:57:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9835218073d420346d48ef0138613528ae0a7189","revisions":{"3d1931a37a2f15a61f1c8fd45c808679f3134043":{"kind":"REWORK","_number":1,"created":"2025-07-30 07:45:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/956132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/956132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/956132/1"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:44:18.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:45:36.000000000","tz":120},"subject":"DNM: Cover more task related things with type hints","message":"DNM: Cover more task related things with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Ib806217076ae85d1744301b381b17f1c7556e5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3d1931a37a2f15a61f1c8fd45c808679f3134043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3d1931a37a2f15a61f1c8fd45c808679f3134043"}]},"branch":"refs/heads/master"},"17912116c0e8b7b38eedca686a62fac75b7088e0":{"kind":"REWORK","_number":2,"created":"2025-07-30 09:20:41.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/956132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/956132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/956132/2"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:44:18.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 09:20:31.000000000","tz":120},"subject":"DNM: Cover more task related things with type hints","message":"DNM: Cover more task related things with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Ib806217076ae85d1744301b381b17f1c7556e5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/17912116c0e8b7b38eedca686a62fac75b7088e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/17912116c0e8b7b38eedca686a62fac75b7088e0"}]},"branch":"refs/heads/master"},"a0f29e62c3dd41fa418f440ddd2caf31c2a4b5d2":{"kind":"REWORK","_number":3,"created":"2025-07-30 10:00:54.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/956132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/956132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/956132/3"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:44:18.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 10:00:36.000000000","tz":120},"subject":"Cover more task related things with type hints","message":"Cover more task related things with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Ib806217076ae85d1744301b381b17f1c7556e5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a0f29e62c3dd41fa418f440ddd2caf31c2a4b5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a0f29e62c3dd41fa418f440ddd2caf31c2a4b5d2"}]},"branch":"refs/heads/master"},"738488025227058f93817efaf54c299a8e21b4dc":{"kind":"REWORK","_number":4,"created":"2025-07-30 10:01:32.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/956132/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/956132/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/956132/4"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:44:18.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 10:01:26.000000000","tz":120},"subject":"Cover more task related things with type hints","message":"Cover more task related things with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Ib806217076ae85d1744301b381b17f1c7556e5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/738488025227058f93817efaf54c299a8e21b4dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/738488025227058f93817efaf54c299a8e21b4dc"}]},"branch":"refs/heads/master"},"80d76621c1d66d0a6c5ffec95f4aae3b9f73c0f5":{"kind":"REWORK","_number":5,"created":"2025-07-30 10:52:07.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/956132/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/956132/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/956132/5"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:44:18.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 10:51:56.000000000","tz":120},"subject":"Cover more task related things with type hints","message":"Cover more task related things with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Ib806217076ae85d1744301b381b17f1c7556e5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/80d76621c1d66d0a6c5ffec95f4aae3b9f73c0f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/80d76621c1d66d0a6c5ffec95f4aae3b9f73c0f5"}]},"branch":"refs/heads/master"},"9835218073d420346d48ef0138613528ae0a7189":{"kind":"REWORK","_number":6,"created":"2025-07-30 11:26:25.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/32/956132/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/32/956132/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/32/956132/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/32/956132/6"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 07:44:18.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 11:26:00.000000000","tz":120},"subject":"Cover more task related things with type hints","message":"Cover more task related things with type hints\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Ib806217076ae85d1744301b381b17f1c7556e5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9835218073d420346d48ef0138613528ae0a7189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9835218073d420346d48ef0138613528ae0a7189"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
