)]}'
{"id":"openstack%2Frally~956146","triplet_id":"openstack%2Frally~master~Id7a8193be2fb76a00bc9fab0c95b1ff551c61901","project":"openstack/rally","branch":"master","topic":"remove-long-deprecated-functions","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2025-07-30 11:26:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"Id7a8193be2fb76a00bc9fab0c95b1ff551c61901","subject":"Remove several deprecated methods from rally.common.utils","status":"MERGED","created":"2025-07-30 10:13:48.000000000","updated":"2025-07-30 12:06:45.000000000","submitted":"2025-07-30 12:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956146-remove-long-deprecated-functions","meta_rev_id":"9f419f741b240856322954c28f8d917e351bdc07","_number":956146,"virtual_id_number":956146,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-30 12:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-07-30 11:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-07-30 11:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 10:55:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"affbada999df95e0f18239940a0262b1f7dda85e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 10:13:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"787a0f14e34ce70e88a7034ad4f076462fd8993c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 10:14:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bcd2439451d08d547c2773f1e133c5f9b01fbc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 10:55:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/066b42e3d13243fabda98f8747447180\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06586c9ac6764ccfbde8b215309089ff : SUCCESS in 4m 09s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7a3740331cff47de9ed235b3e40bff4f : SUCCESS in 4m 38s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/6976bbccf56d418aaac95ab11f0cafd4 : SUCCESS in 3m 49s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/855a54bee75f41f2b97fd4fffc41a21e : SUCCESS in 4m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/913746702fc04b24ba12dd721c176140 : SUCCESS in 4m 05s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/212441e1b9964d6da324afa10c0ebcb3 : FAILURE in 1m 58s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/992148b47bfe413ab34ca23de2a6d655 : FAILURE in 2m 58s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e2bea8ea5aed474c8bb88b4e93b86e1e : FAILURE in 2m 38s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/76be191b752b436b935f6ef726617f6e : FAILURE in 2m 01s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/0dfd8d3826f5466790f814763eb2811c : SUCCESS in 1m 39s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/917654c4d0eb4941a2a0effb2fa5fe9f : SUCCESS in 4m 28s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2bcba861a03f4e1180a7a63695d84329 : SUCCESS in 6m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccac38d890e8490e81b99c90abf93ded : SUCCESS in 5m 39s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8006871877964acba10dce6a740a2b2b : SUCCESS in 6m 12s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/1c2829655a2f469d812461e6b9ea9add : SUCCESS in 5m 12s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/a32a5bfaa2f7473a893510d0d2f9c110 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"600f2913ceac859bb573c52f95767403c1695923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 11:26:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbb27b99f6ba80fb78f026ca3539dc1b879abe3e","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-07-30 11:26:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83357eafe01a8ac97209098ceb71e45160e43d64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:53:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff370809ecf343a68880af83f9495cb8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1475f3ab7c8e4bab99b966f23426c281 : SUCCESS in 6m 01s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/3e616417ca634b989ce1131e1a743792 : SUCCESS in 5m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4c1ad0f6f3bb49ddae6dc5b4eb5f922d : SUCCESS in 5m 24s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/694b7869b5a94ad798382f1256a2057c : SUCCESS in 1m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6acc2446c147e9a8475cf85e632539 : SUCCESS in 2m 30s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/3293f084037e4171b072676dec6018e5 : SUCCESS in 2m 43s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/0ee68c5ed8fd4c03ad19db4a5718c486 : SUCCESS in 3m 03s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/81803ea1ade146dab32eafe251cbfaae : SUCCESS in 2m 41s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/86154083925c472bb978f3715560867f : SUCCESS in 2m 42s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/897366d9412b499fa4b0e3258206fc2a : SUCCESS in 3m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b208f5c2928b420da7bd37be11d65ebf : SUCCESS in 6m 39s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/30d4282faa424984ba3bb8940444b779 : SUCCESS in 6m 21s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d77489eeb4648d3860a29adb8fa55ca : SUCCESS in 5m 26s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b613dc54a38047e6954b47794d08b137 : SUCCESS in 6m 03s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/ce2863310f104cc4bfde3a351528d936 : SUCCESS in 6m 40s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6be8f4464e414a16828cfd1c5864bfcd : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"5929bbd54354c3fdb652d6d51f94638c6a4201cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 11:53:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9fbfb70b6dc82d57fcefba61d9d9e5b9f983e8b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 12:06:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94497ee8091f4b05b0b128722f9b19c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5ba3b40fcd488d9a458e5c9ddc1e68 : SUCCESS in 5m 27s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8be62c7315fe4dd6a4d0067c1b03cc08 : SUCCESS in 5m 02s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/bb61ee7597064e2e847c2d71dae167d8 : SUCCESS in 3m 31s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/434aa3bd571f4df2b7d9781c782b5b92 : SUCCESS in 4m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02adc2f3e2442e9952c797fcfd45375 : SUCCESS in 2m 25s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/3edb2c784a69421da098f3d6f4aaf20f : SUCCESS in 3m 06s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/7fb3464e90584af5a5e16afea1e06783 : SUCCESS in 3m 55s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/96e9e28eb2ed40f9af4b665e16be0d94 : SUCCESS in 3m 06s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/d3b815f140f245efa0add738619310d0 : SUCCESS in 3m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/032b4f4b34b54b9d988e47a281cc3b88 : SUCCESS in 2m 26s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7e9095d3ceeb4d908ac93675bce6160f : SUCCESS in 7m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6420d5bf0256405fa9ffe44f204c7020 : SUCCESS in 5m 25s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03ac4ce1bf64a1b8528bd271f5e2ac3 : SUCCESS in 5m 30s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41f3e73be6ef4b89a28d58de3d3876ba : SUCCESS in 6m 07s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/7686790225274ae9971fa0b387e04c09 : SUCCESS in 6m 10s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/89282e842d774a48b13c2065dd7a11fc : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9f419f741b240856322954c28f8d917e351bdc07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 12:06:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1e22f1d46f9a50b1ac47d2b80efead1d28c05fb","revisions":{"5ad676a750195ac3c05ecb0286178527a0d3e85c":{"kind":"REWORK","_number":1,"created":"2025-07-30 10:13:48.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/46/956146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/46/956146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/956146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/956146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/956146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/46/956146/1"}}},"commit":{"parents":[{"commit":"3ca8e862883e8003c11fe6a6f94354babd5fca1c","subject":"Cover task plugins with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ca8e862883e8003c11fe6a6f94354babd5fca1c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 10:10:36.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 10:13:31.000000000","tz":120},"subject":"Remove several deprecated methods from rally.common.utils","message":"Remove several deprecated methods from rally.common.utils\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Id7a8193be2fb76a00bc9fab0c95b1ff551c61901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5ad676a750195ac3c05ecb0286178527a0d3e85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5ad676a750195ac3c05ecb0286178527a0d3e85c"}]},"branch":"refs/heads/master"},"c1e22f1d46f9a50b1ac47d2b80efead1d28c05fb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-30 11:26:47.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/46/956146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/46/956146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/956146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/956146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/956146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/46/956146/2"}}},"commit":{"parents":[{"commit":"9835218073d420346d48ef0138613528ae0a7189","subject":"Cover more task related things with type hints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9835218073d420346d48ef0138613528ae0a7189"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 10:10:36.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-07-30 11:26:47.000000000","tz":0},"subject":"Remove several deprecated methods from rally.common.utils","message":"Remove several deprecated methods from rally.common.utils\n\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\nChange-Id: Id7a8193be2fb76a00bc9fab0c95b1ff551c61901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c1e22f1d46f9a50b1ac47d2b80efead1d28c05fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c1e22f1d46f9a50b1ac47d2b80efead1d28c05fb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
