)]}'
{"id":"openstack%2Frally~956994","triplet_id":"openstack%2Frally~master~I893f2655311e2312341cc93637046bb08f8fb7ef","project":"openstack/rally","branch":"master","topic":"task-v2-samples","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2025-08-10 10:43:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I893f2655311e2312341cc93637046bb08f8fb7ef","subject":"Port all samples to task v2 format","status":"MERGED","created":"2025-08-10 09:52:00.000000000","updated":"2025-08-10 10:56:45.000000000","submitted":"2025-08-10 10:56:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956994-task-v2-samples","meta_rev_id":"681ab1f9ab50b2f41a34194dc992206b1e57fb0e","_number":956994,"virtual_id_number":956994,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-10 10:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-08-10 10:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-08-10 10:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-10 10:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"978c052388b0a602ece1471b0fed48df8dd12043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-10 09:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d3bd021aeec0f47ddb3f81de157655820ab01bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-10 09:56:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d2c963a194cd48ae05c839886a8f4673d324dc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-10 09:59:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"93fdbf940d51da605d596652f4ac58ce3c65793b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 10:19:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb3d975a1d45497691cd30d1023bf01e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cfbe84f16a4e1eb4043becfa719e6e : SUCCESS in 4m 24s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2b78014b032a4ba08968797e739e0595 : SUCCESS in 4m 17s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4cc0123f939b40fa9026988dcc2f1717 : SUCCESS in 3m 24s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/6a0fc3235c1b4d0fa73cf2ba53cdf3e8 : SUCCESS in 1m 40s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d65eb8649ad4dc59b9ca65c73a1144e : SUCCESS in 2m 30s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/0d37ed743d694d4f9b207a22a8dd80d5 : FAILURE in 2m 45s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/c8686c635fdf4eb481030d399669d68b : FAILURE in 2m 32s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/bde069e0a4b5479baa1604b7fb2e701a : FAILURE in 2m 58s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/c716ba320eff45c5a092e7579803d811 : FAILURE in 2m 51s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8dbf3f39f1d2477c80053895bf4b1ba8 : SUCCESS in 2m 47s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d8f34edec1a44b7691c0959e8acda7d5 : SUCCESS in 8m 09s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e64a71fef2034268b7baeb130b8b30c8 : SUCCESS in 6m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db254e4f4205433d948be173026f3e75 : SUCCESS in 5m 51s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/057740c932be4a11ab9a00f02c6f2b6f : SUCCESS in 5m 40s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/f69919843dd44e939e0234c17493c433 : SUCCESS in 6m 18s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7e5e4e144ece4ed98833f1bac0a5d85f : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"911a17073aaa9415296d94c6b147a8e835a357cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-10 10:28:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"945e1cef138c55f13d51871db307e4cfad4ad4d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 10:38:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5f0b524d3aa4f6095a3674fd002068e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f79d30893ef4d778cf9f58c767b48e1 : SUCCESS in 3m 54s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b68fd4eea078465d8eb289cf7db881a7 : SUCCESS in 4m 08s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2daaef6cdda343d2b231098ba658e07c : SUCCESS in 1m 55s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/84a49d60e5f6473ea385e31caff74cfb : SUCCESS in 1m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/465a6cac193043948e3e95925196741d : SUCCESS in 1m 29s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/cc600dea89d64141be90c68d2fbd1677 : SUCCESS in 2m 07s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/eabd7a3b2cf044349394f5604260b633 : SUCCESS in 1m 39s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/54391c9819924626bd3b37f5a213cffa : SUCCESS in 2m 07s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/17bb0648a7c04e61812b254f769c655b : SUCCESS in 2m 55s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/4567404cf9e649c4ac3a4db5060c332d : SUCCESS in 1m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6349f623ce3940d9a86825a672f3137c : SUCCESS in 5m 54s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ba6b5cb565904e548d5ca92770c603c0 : SUCCESS in 6m 30s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9dfc98d3a0d47879195020aa366c494 : SUCCESS in 4m 40s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd9a8cf7919c4fd0951367f452ddc8e6 : SUCCESS in 4m 58s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/2785dd8888a34a25ac3cc4b06dc24403 : SUCCESS in 6m 02s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ec2e44e6b086443480d5a914a8238c8a : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a8e9f986cb9ddd5c8cd621c2845f77d4bdf40400","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-08-10 10:43:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aa53f91b1f0368dd7a47a7ef0a9e45d447884eb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 10:44:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2a89eab3888dae4a0b054f8d041d905581b197fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 10:56:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f91969a156649f5998a0821fd7d37d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfad232925494bd9958d0b24641ed8cb : SUCCESS in 2m 35s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/74c29df8a87945729c8d6e5a6bcd03de : SUCCESS in 5m 27s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ac5b988488204c66a63a34619e520ea0 : SUCCESS in 3m 44s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/669389c4b51f42d19ebf830617be30f4 : SUCCESS in 1m 45s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b4f8dbff468473e9459b125132eca0d : SUCCESS in 2m 29s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/001a546cd3fd4f44b9bdc516f12b6cfc : SUCCESS in 2m 07s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/3607fc5204854221937b49efd867aaf4 : SUCCESS in 2m 04s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/c2f8351812d9474bb8a673104e8264fc : SUCCESS in 1m 39s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/64184eceedc14ac1a56acec6d1eda675 : SUCCESS in 1m 51s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/59c4f37fb35b40e79ed60d8104df743f : SUCCESS in 1m 35s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/5af1eebe4e5d4b4b96be9fef193a3a97 : SUCCESS in 4m 51s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/efad12d7f27c48ff9f2ffd7660c8db32 : SUCCESS in 5m 08s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6b1c237180242d6814e3cb7eb854a2e : SUCCESS in 4m 50s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67262308f411419782c3236d073ced90 : SUCCESS in 4m 48s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/d34abe7c966340d3b5cc811f0f044418 : SUCCESS in 6m 15s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bde6b66df8bb43faa75bb01909c3ffb1 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":4},{"id":"681ab1f9ab50b2f41a34194dc992206b1e57fb0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-10 10:56:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6565717213fa7b87f383872d5d639b4d87c5faf2","revisions":{"cf309b0a0e8b0b27f1084a605cc2c762274db9f3":{"kind":"REWORK","_number":1,"created":"2025-08-10 09:52:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/94/956994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/94/956994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/94/956994/1"}}},"commit":{"parents":[{"commit":"cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9","subject":"[type-hints] Set config of context and hooks as Any","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:51:15.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:51:58.000000000","tz":120},"subject":"Port all samples to task v2 format","message":"Port all samples to task v2 format\n\nChange-Id: I893f2655311e2312341cc93637046bb08f8fb7ef\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cf309b0a0e8b0b27f1084a605cc2c762274db9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cf309b0a0e8b0b27f1084a605cc2c762274db9f3"}]},"branch":"refs/heads/master"},"9d17c33d3edda2102368f9629dd09bf47adc043c":{"kind":"REWORK","_number":2,"created":"2025-08-10 09:56:08.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/94/956994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/94/956994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/94/956994/2"}}},"commit":{"parents":[{"commit":"cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9","subject":"[type-hints] Set config of context and hooks as Any","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:51:15.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:56:00.000000000","tz":120},"subject":"Port all samples to task v2 format","message":"Port all samples to task v2 format\n\nChange-Id: I893f2655311e2312341cc93637046bb08f8fb7ef\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9d17c33d3edda2102368f9629dd09bf47adc043c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9d17c33d3edda2102368f9629dd09bf47adc043c"}]},"branch":"refs/heads/master"},"67177dd4033b82b6f3bd9c4eb8ba14454a873a02":{"kind":"REWORK","_number":3,"created":"2025-08-10 09:59:07.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/94/956994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/94/956994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/94/956994/3"}}},"commit":{"parents":[{"commit":"cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9","subject":"[type-hints] Set config of context and hooks as Any","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:51:15.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:58:37.000000000","tz":120},"subject":"Port all samples to task v2 format","message":"Port all samples to task v2 format\n\nChange-Id: I893f2655311e2312341cc93637046bb08f8fb7ef\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/67177dd4033b82b6f3bd9c4eb8ba14454a873a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/67177dd4033b82b6f3bd9c4eb8ba14454a873a02"}]},"branch":"refs/heads/master"},"6565717213fa7b87f383872d5d639b4d87c5faf2":{"kind":"REWORK","_number":4,"created":"2025-08-10 10:28:18.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/94/956994/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/94/956994/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/956994/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/94/956994/4"}}},"commit":{"parents":[{"commit":"cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9","subject":"[type-hints] Set config of context and hooks as Any","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cd921eafc4cd0ba9aaeb8efa4103a387f2fdc3a9"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 09:51:15.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-08-10 10:28:11.000000000","tz":120},"subject":"Port all samples to task v2 format","message":"Port all samples to task v2 format\n\nChange-Id: I893f2655311e2312341cc93637046bb08f8fb7ef\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6565717213fa7b87f383872d5d639b4d87c5faf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6565717213fa7b87f383872d5d639b4d87c5faf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
