)]}'
{"id":"openstack%2Frally~959346","triplet_id":"openstack%2Frally~master~I812f2730a830feca0a9248153fd168e267849087","project":"openstack/rally","branch":"master","topic":"un-release","hashtags":[],"change_id":"I812f2730a830feca0a9248153fd168e267849087","subject":"Get back to pbr","status":"MERGED","created":"2025-09-03 16:07:41.000000000","updated":"2025-09-03 16:36:05.000000000","submitted":"2025-09-03 16:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959346-un-release","meta_rev_id":"45f96b742715041339c18d9dc6d9bbb353139835","_number":959346,"virtual_id_number":959346,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 16:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-09-03 16:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-09-03 16:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 16:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1b6fd36cada032ac3fe9fca8f05b392906e0ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-09-03 16:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2da5a62c19f49b698f331d009367ef7bead44590","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-09-03 16:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95b5396884bd1bdf447ad69e2b45be70aba6f5bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb2a566fb6c34cde8cde40f9d418500b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed5240c3a2954edb91bb39a094ad17ae : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b5180a34c94801a83d280b37dd89b1 : SUCCESS in 2m 53s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/52ed9ec9333140ea93cfffc4a9eb6671 : SUCCESS in 5m 09s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/236d47ac7c1f4854bcbf292f9fae7bad : SUCCESS in 4m 32s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/e65b000c9cc14763859243f576a37d8a : SUCCESS in 3m 31s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a58bebda6e6349d49a1e3f412b49eccb : SUCCESS in 4m 44s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/40b72c16173644348ac9b79d80b2ea6a : SUCCESS in 3m 49s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/eec040f29379416392a1dec8d60f4c7f : SUCCESS in 3m 42s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/86d25ec5c46c4c24ad864fd254ef1738 : SUCCESS in 3m 29s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/b9b696e79915439eadbb12798a89687b : SUCCESS in 5m 03s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/3d25e83c9d5448e4872bc8694f1b6b4f : SUCCESS in 3m 57s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/39be339f0a8f4b549f9794d998a9b797 : SUCCESS in 4m 58s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f386eca4270e48b0912689976fba6036 : SUCCESS in 7m 06s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe94faf4be5d4e31993e8c2211c58586 : SUCCESS in 5m 40s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a170f6dcddc744f69e8d7016e5d4fbfd : SUCCESS in 5m 42s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/7276285daa1140d3beceb402bc31ecff : SUCCESS in 5m 59s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/302eb6311fe84b6b9d06569c0752b2ac : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e6605e06ea03b7a3e3c9823fee5a8432869f7165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:19:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fc279b805019a7a2091f1baf233c81879d3f414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:36:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8953bfc4f9e54b6f988489e0d80904c7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bccc2dd96b1642fa81832b0a4891a8ef : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0de13d9705748cc8a1846bfc1e3adc0 : SUCCESS in 5m 10s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c8a6f915c58347b094a5d0c25f2b9092 : SUCCESS in 4m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4f936d88f9e147238dbe9c0548f38ae6 : SUCCESS in 2m 44s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/25adf0af66b943e784c34a950c7fcd0e : SUCCESS in 4m 31s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e95656daad487397c8672e6a5dc222 : SUCCESS in 3m 38s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/08e40e155fe345d1aefecbb61127f163 : SUCCESS in 3m 20s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/d8899b01289546d5826c75b3777fcc62 : SUCCESS in 3m 26s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/157f7c4954804a7e89af2c5a807a24a2 : SUCCESS in 3m 50s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/9d869681f0a34f998d57f74f065ba14d : SUCCESS in 3m 19s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/06d1ee53878249bba28d8cbadbd75418 : SUCCESS in 3m 15s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/cbded55f51ff4e74bb57341e72c1d391 : SUCCESS in 6m 46s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2418d972a0ea4024972e5f33222f1242 : SUCCESS in 6m 37s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a561899131224f659e25d03663155bca : SUCCESS in 6m 49s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e788ac21984429dae0a08d3f0c61555 : SUCCESS in 8m 06s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/8a8d6070b87f4688905cbdb2b16359b5 : SUCCESS in 7m 28s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/81a75b1b4db14900afddeb3917e42a53 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"45f96b742715041339c18d9dc6d9bbb353139835","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:36:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fa6c6aa24c2d7bd2a0b39fb58479cb59615554c","revisions":{"1fa6c6aa24c2d7bd2a0b39fb58479cb59615554c":{"kind":"REWORK","_number":1,"created":"2025-09-03 16:07:41.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/46/959346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/46/959346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/959346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/959346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/959346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/46/959346/1"}}},"commit":{"parents":[{"commit":"5e947518c311b97266427df5f19b64f664e93029","subject":"Revert \"Propose Rally 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5e947518c311b97266427df5f19b64f664e93029"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-09-03 16:02:42.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-09-03 16:07:07.000000000","tz":120},"subject":"Get back to pbr","message":"Get back to pbr\n\nWe need it to make rally be compatible with releases software.\n\nThis also adds some type hings for migrations - let\u0027s move forward with\nit.\n\nChange-Id: I812f2730a830feca0a9248153fd168e267849087\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1fa6c6aa24c2d7bd2a0b39fb58479cb59615554c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1fa6c6aa24c2d7bd2a0b39fb58479cb59615554c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
