)]}'
{"id":"openstack%2Frally~959347","triplet_id":"openstack%2Frally~master~I8afc893f4e6d4c766836edc8cb2336332f9806eb","project":"openstack/rally","branch":"master","topic":"un-release","hashtags":[],"change_id":"I8afc893f4e6d4c766836edc8cb2336332f9806eb","subject":"Propose Rally 5.0.0 release","status":"MERGED","created":"2025-09-03 16:09:46.000000000","updated":"2025-09-03 16:48:21.000000000","submitted":"2025-09-03 16:48:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959347-un-release","meta_rev_id":"0a20882da6b34e385c3edbf9a744410fd02d4dbf","_number":959347,"virtual_id_number":959347,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 16:48:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-09-03 16:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-09-03 16:10:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 16:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d3f71d8faeb23edd1f8c6c055322578feee8807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-09-03 16:09:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c58aacd48d524468d705ce95c1f67900eaeeb70c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-09-03 16:10:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eac889da23b080d0b7dd43f3f9e1140062c8743d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:36:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/114769fdcd2b48dcaef95ff5c57f8212\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dc490332ea445090296cda135865f4 : SUCCESS in 4m 56s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f1b4b4b2051e4af593f3e9b826492ed3 : SUCCESS in 4m 15s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f2ec37aaaed7447299e93f723cf5d693 : SUCCESS in 4m 37s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/cc2cf48a8d3f40acb349c8faaf6229ba : SUCCESS in 3m 40s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fc1ccae7c045c28cefe9dfbf131fe8 : SUCCESS in 3m 34s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/00bc175731564dd68091729bd5948704 : SUCCESS in 3m 38s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8c1f2fea572043b0922c900180c60e08 : SUCCESS in 4m 00s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/fe301ba4c4bf47dab2a318a596258251 : SUCCESS in 5m 08s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/89849b51e3e04ca1bdb7dc31cc6ddbba : SUCCESS in 3m 49s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/76c444e80dac499dadc30268ad3cd71b : SUCCESS in 3m 17s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ca3c94595e9c46f8850e72f6cecaeba2 : SUCCESS in 8m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/00968031b1c44b818f6c01f07bcb92cd : SUCCESS in 6m 59s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3e828c185af4db48f45fda95dd35b7c : SUCCESS in 5m 41s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eea0afe26b8b4d0688bb0a54eb881c89 : SUCCESS in 6m 16s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/c649058944a546fca67ffabba4af0da2 : SUCCESS in 6m 12s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/af6792eb921a4f04b1a7e8c2c13ed36c : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b8d391a51423df8039e3bc0a518e1a36665384c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:36:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7476859a3c370e5caeee5b1c1c269b953b2425a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:48:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9094fc41e6b449eb8f97bb769ae96ccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ecba391c9714cdd87e9dcabb8d1392e : SUCCESS in 2m 44s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/67bcd640a5a64f8aad0d71b12278a104 : SUCCESS in 3m 05s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cb4b1925347e4ea7a0b67357956bd62a : SUCCESS in 2m 42s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/b0c2e996cda94def85dc51360541417e : SUCCESS in 2m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a4f85fc3cc49c9b11c9144bace66ae : SUCCESS in 2m 14s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/ff4db5198562420cacaa830db0581bc3 : SUCCESS in 2m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/02fd41a989f44d83bef7dccfd132e921 : SUCCESS in 2m 25s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/d44bf38317454cffab145809bcf1ac8f : SUCCESS in 4m 17s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/c00d6645af824f05bcd6812c02f38cdb : SUCCESS in 3m 15s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/9831814b5063441db73be6ec86e09802 : SUCCESS in 4m 36s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/82d0c7cb84c94f2e919c0942bf1fb460 : SUCCESS in 4m 54s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/644b1820aea646fd914bbfaaaa11a3f4 : SUCCESS in 6m 40s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eacaaa780ada43a28e36cef462143581 : SUCCESS in 4m 48s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb86836b68f145ec922e5e4f55078aff : SUCCESS in 6m 22s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/70dbfddbe5b247228630464fcb6c7366 : SUCCESS in 7m 21s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7a2012f473ac4b1abfdfb90c6dd3add2 : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0a20882da6b34e385c3edbf9a744410fd02d4dbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 16:48:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06e62af58c38ad3161398f1f99daa3d78d6d8582","revisions":{"06e62af58c38ad3161398f1f99daa3d78d6d8582":{"kind":"REWORK","_number":1,"created":"2025-09-03 16:09:46.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/47/959347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/47/959347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/959347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/959347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/959347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/47/959347/1"}}},"commit":{"parents":[{"commit":"1fa6c6aa24c2d7bd2a0b39fb58479cb59615554c","subject":"Get back to pbr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1fa6c6aa24c2d7bd2a0b39fb58479cb59615554c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-09-03 16:09:30.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2025-09-03 16:09:38.000000000","tz":120},"subject":"Propose Rally 5.0.0 release","message":"Propose Rally 5.0.0 release\n\nChange-Id: I8afc893f4e6d4c766836edc8cb2336332f9806eb\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/06e62af58c38ad3161398f1f99daa3d78d6d8582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/06e62af58c38ad3161398f1f99daa3d78d6d8582"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
