)]}'
{"id":"openstack%2Frally~971779","triplet_id":"openstack%2Frally~master~I7c1f9ac64be31985271719dbeee898e140145d47","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-29 21:02:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c1f9ac64be31985271719dbeee898e140145d47","subject":"Fix pep8 error","status":"MERGED","created":"2025-12-24 16:53:48.000000000","updated":"2025-12-29 21:02:51.000000000","submitted":"2025-12-29 21:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971779","meta_rev_id":"5d0749c43084fe81458ecd188a6f62b649825d37","_number":971779,"virtual_id_number":971779,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-29 21:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2025-12-29 20:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2025-12-29 20:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-24 17:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-29 20:48:34.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"9df955133098cab4610c79ae1ffd3168eb44d15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-24 16:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d1028efc2568f7a0f60e6bde53ae5254f6cedc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 17:13:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6d6aad50208462ba24c5f3053729717\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3a2afcf94a4ecf827bbb6bec48a7a2 : SUCCESS in 4m 43s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/211eab433d9949d28132f881add42395 : SUCCESS in 3m 59s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/487fd9bfa00849c88978386ef83d3a55 : SUCCESS in 3m 47s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/db0aac403e514731a5c0320d9c669d5a : SUCCESS in 3m 16s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c150f46ef84b4a8b6f3efb3e6c8a83 : SUCCESS in 2m 16s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b95c3ff118455aad14d019e092fb97 : SUCCESS in 3m 23s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/d7677724898a41c6b32d5bf90cd9672c : SUCCESS in 3m 32s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/17c011c474ea4d9080a5fea1e286920d : SUCCESS in 3m 09s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/b7cb0e008d6c48beaac0480f9007576b : SUCCESS in 3m 45s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/c26615d207b74910b1ad5e024d3795e5 : SUCCESS in 2m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/5dd05290b8cb4b0a9a045bae3f9d2ff4 : SUCCESS in 5m 30s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/217398a77dbd467581223fb427460d9d : SUCCESS in 6m 13s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/337a8358e1c343af9674449dcbd561e1 : SUCCESS in 5m 28s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb92d3bc164e422689c3841f731714d2 : SUCCESS in 6m 24s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/9bef611773fd413ea4db1f961c14d348 : SUCCESS in 5m 32s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/df3c22667ea243ff98cee6c11ac45ddc : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0eed1f528d44c1b00c515af0ff662b8843a6939b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2025-12-29 20:48:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7478a0e72e091327845fbe4ada14ec3f54f513c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 20:48:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c566eeb9f78c9dada6145ef495e2239a62129757","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 21:02:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/378109e4fe5a4facaea893113ddc8116\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19256831b80841258f720b8f1ad0eb66 : SUCCESS in 5m 21s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b5dfe47b9c9d4f65b8fe14bd8c38ec2d : SUCCESS in 5m 35s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/669e965538a54ecf8e3bb6a9fe90d230 : SUCCESS in 3m 47s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a0635808e3834f2bb42311361e32a4d7 : SUCCESS in 2m 49s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2def03ff444a4b9fe16cb6ea0a13f1 : SUCCESS in 3m 10s\n- rally-tox-py39 https://zuul.opendev.org/t/openstack/build/6297bcfbd7254f9e9c61fb764d13a13e : SUCCESS in 3m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/14fd298e06b741b6a082aa47992f188b : SUCCESS in 2m 06s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/c19d7e8482dc441ab587eb3393225c6a : SUCCESS in 2m 46s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/07f449cbac3e4760a8a9ca1828b9a653 : SUCCESS in 4m 04s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/92d0ba4bc40146c38ce8c72a2f975ca2 : SUCCESS in 3m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/0ec596249ddc442fa7c03b539995f721 : SUCCESS in 4m 36s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c333416091f14f98af395b6fd1ebfded : SUCCESS in 6m 04s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8dc6279611243c4afba50babf8a629b : SUCCESS in 5m 12s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c46955767b64fd9b0bf4178c7dd9419 : SUCCESS in 6m 13s\n- rally-install-centos-9s https://zuul.opendev.org/t/openstack/build/e245d787d6eb4e0baff019dbef45ae7c : SUCCESS in 6m 08s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/08817e01cdf5498b937540f224366c04 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5d0749c43084fe81458ecd188a6f62b649825d37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 21:02:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41f9886ac1af3dcfed83e24fa2ad066a56a4aea7","revisions":{"41f9886ac1af3dcfed83e24fa2ad066a56a4aea7":{"kind":"REWORK","_number":1,"created":"2025-12-24 16:53:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/971779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/79/971779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/971779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/971779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/971779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/79/971779/1"}}},"commit":{"parents":[{"commit":"26cdfbe70278e58aa7789f1b6d4283387fd19b11","subject":"Propose rally 5.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/26cdfbe70278e58aa7789f1b6d4283387fd19b11"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-24 16:53:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-24 16:53:38.000000000","tz":540},"subject":"Fix pep8 error","message":"Fix pep8 error\n\nResolve the following error detected by flake8.\n\n```\n./rally/common/opts.py:49:5: F824 `global _registered_paths` is unused:\nname is never assigned in scope\n```\n\nChange-Id: I7c1f9ac64be31985271719dbeee898e140145d47\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41f9886ac1af3dcfed83e24fa2ad066a56a4aea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41f9886ac1af3dcfed83e24fa2ad066a56a4aea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
