)]}'
{"id":"openstack%2Frally~996128","triplet_id":"openstack%2Frally~master~Id5b63c6dc732a2a20915a9eb1d1e712e282df504","project":"openstack/rally","branch":"master","topic":"ruff","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-07-06 17:30:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"Id5b63c6dc732a2a20915a9eb1d1e712e282df504","subject":"Start adopting ruff","status":"MERGED","created":"2026-07-06 16:23:58.000000000","updated":"2026-07-06 18:13:24.000000000","submitted":"2026-07-06 18:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996128-ruff","meta_rev_id":"b9a5e395c79e451d2edbb7221e561b3a349a6f26","_number":996128,"virtual_id_number":996128,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-06 18:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-07-06 17:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-07-06 17:30:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 17:21:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cea43c37298218fa1d981cec395f5bf5c90da230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 16:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"145ea22e698b56f94ab132b52d5a700a3222e3f4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 16:25:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9da94ffec7865c66b8a67f619a884d17ed148e0d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 17:13:05.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"92d4c1ca0585f0f760d5220a122b8a08612f70d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 17:21:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be33876939fb4b0a8cf1d605cf57229c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c0063697b2d4008a3230d4ee5cda487 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a6e9700e739433b993fd7176c135a85 : SUCCESS in 5m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/61555c41212b40cfa1e509ef26b62948 : SUCCESS in 4m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cf9dccf76b2040738d14a4c3b02798d3 : SUCCESS in 4m 07s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/0695ce3e44ae4bc9b5fc1a75b3d1f1ba : SUCCESS in 3m 30s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8fd83ce5d142baa58196e4e2e90145 : SUCCESS in 2m 56s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b2e33ecc3f4948d2902c5d9b03fecb49 : SUCCESS in 2m 31s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/59201375ec844c3ca9ee3d774b14fa78 : SUCCESS in 2m 45s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/caaa37729b7c40848d9c28883a82693c : SUCCESS in 3m 00s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/31a3529e29544ec0b4f6e711ee2c70a3 : SUCCESS in 2m 14s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/397ea734ca9641118ab3994f473e4904 : SUCCESS in 3m 03s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3371867e0b474e0d870b5087156b7780 : SUCCESS in 4m 54s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e5eef94e2882486db3346ab7c90d98ac : SUCCESS in 6m 09s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d44b5101d32740d0af7a20756d58ae8a : SUCCESS in 5m 20s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1df27ddb0c94b9caf4ee1ce9138305b : SUCCESS in 5m 43s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/d62b33f468b648799a6a9822f8f01ce4 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"968d161015fec99364bd736b9e7b18bc9149ad3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 17:25:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e3fb6230efd660fcfac84af3bb38bf1ad8e2adb0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 17:26:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"126c770464b0cea00e15d0d9ae9fbe477fdcbfa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 17:30:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fcf292d0ce803e0ffb1efa39132a20a32de48709","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-06 17:30:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"95d1002b551b0056c2b633339acd6100427d1616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 17:51:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a80f58ba6f47fa86c78db4f1e895c2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/40fe2e07eaf24929951004126a7a8c3f : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c511924074a01970a864a7791257e : SUCCESS in 7m 19s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/961788ba94354e11a433baf7c25fc93d : SUCCESS in 5m 24s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/3c1d5b0492414379b5edc4bb27bd8a8d : SUCCESS in 4m 54s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/f6cdab82058b47d5a720d3f776928b12 : SUCCESS in 3m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ea083c66be4254b3754d7cafc374d0 : SUCCESS in 3m 07s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/d3f6fd4be3c54b419063d55c745844c8 : SUCCESS in 3m 14s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/ceb853c20ac74910a5dcdbb99956495b : SUCCESS in 3m 00s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/c6ca86724f9e410eaac90b5b89a503ed : SUCCESS in 5m 04s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/d0be377240a746aab418c70919f97737 : SUCCESS in 4m 06s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/058eceee4e7c482da22a5f15113030f2 : SUCCESS in 2m 48s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b1cf148f45bd4533b17e6b1c6c5b564c : SUCCESS in 5m 53s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d29ef27c03ce4bf5bc3e025037c2bbba : SUCCESS in 8m 50s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2084fc62b221440b9ee8b07f65f46578 : SUCCESS in 6m 18s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a85695a0ceb146889181e5d70da91e17 : SUCCESS in 7m 06s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7b5ea5d0e18b4cb7874acefa82d2f732 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":3},{"id":"8d981709f5db26deaf670bb0c1b467113372d865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 17:51:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"154f93f8891d3bf960b81dc63c96079f1d2a9c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 18:13:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb92f77855944202801bfd715e0d8b74\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36986ba72b884097a6f99620891c4368 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b903459bf066459396803260aa6f1c5f : SUCCESS in 3m 44s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9ca6922e69864633b65c98ecc6b59860 : SUCCESS in 5m 59s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/319cc10e716848758cfda124a02137a2 : SUCCESS in 3m 32s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/eed7c715ae5f4771ac2948be97b12c51 : SUCCESS in 3m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f99a205fc4242efa5b2cf7e0e3e6a8d : SUCCESS in 2m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/7d0496858fa64cbfbe8853f36da0bfff : SUCCESS in 3m 59s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2289748f8ab94005956cfdc7457f5ee7 : SUCCESS in 2m 48s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/77e6de6ab9c449fb81812cd3349a7964 : SUCCESS in 3m 41s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/529a1b4287d24e908d0b18ec9efcaa99 : SUCCESS in 4m 53s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ece92f50cdd847029b21d705c1d2ee42 : SUCCESS in 2m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7621d67f12bd46158a3253c028089f24 : SUCCESS in 6m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/33513df631d2411ab78cbafc0a747ea0 : SUCCESS in 7m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5070ce1e54045f482ea2da91364da62 : SUCCESS in 5m 57s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80c856fe83cd447595c4f0f75416190c : SUCCESS in 5m 58s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/acd955ba47cc405b9715435207de108b : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":3},{"id":"b9a5e395c79e451d2edbb7221e561b3a349a6f26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 18:13:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18d6c2639764f8a6e560bf27950698825bd0a47c","revisions":{"966d616659588212aab78c616eb0c026d33afb69":{"kind":"REWORK","_number":1,"created":"2026-07-06 16:23:58.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/996128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/996128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/996128/1"}}},"commit":{"parents":[{"commit":"9342c2359532b62992fa6adb6ab643900616832e","subject":"tox: Remove redundant environment variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9342c2359532b62992fa6adb6ab643900616832e"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-06 16:23:41.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-06 16:23:56.000000000","tz":120},"subject":"Start adopting ruff","message":"Start adopting ruff\n\nChange-Id: Id5b63c6dc732a2a20915a9eb1d1e712e282df504\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/966d616659588212aab78c616eb0c026d33afb69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/966d616659588212aab78c616eb0c026d33afb69"}]},"branch":"refs/heads/master"},"b10b0e4b07c5ea965ed40f7157268d003edd2ead":{"kind":"REWORK","_number":2,"created":"2026-07-06 17:25:54.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/996128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/996128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/996128/2"}}},"commit":{"parents":[{"commit":"9342c2359532b62992fa6adb6ab643900616832e","subject":"tox: Remove redundant environment variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9342c2359532b62992fa6adb6ab643900616832e"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-06 16:23:41.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-06 17:25:38.000000000","tz":120},"subject":"Start adopting ruff","message":"Start adopting ruff\n\nChange-Id: Id5b63c6dc732a2a20915a9eb1d1e712e282df504\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b10b0e4b07c5ea965ed40f7157268d003edd2ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b10b0e4b07c5ea965ed40f7157268d003edd2ead"}]},"branch":"refs/heads/master"},"18d6c2639764f8a6e560bf27950698825bd0a47c":{"kind":"REWORK","_number":3,"created":"2026-07-06 17:30:14.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/28/996128/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/28/996128/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/28/996128/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/28/996128/3"}}},"commit":{"parents":[{"commit":"9342c2359532b62992fa6adb6ab643900616832e","subject":"tox: Remove redundant environment variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9342c2359532b62992fa6adb6ab643900616832e"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-06 16:23:41.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-06 17:30:02.000000000","tz":120},"subject":"Start adopting ruff","message":"Start adopting ruff\n\nChange-Id: Id5b63c6dc732a2a20915a9eb1d1e712e282df504\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/18d6c2639764f8a6e560bf27950698825bd0a47c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/18d6c2639764f8a6e560bf27950698825bd0a47c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
