)]}'
{"id":"openstack%2Frally~996706","triplet_id":"openstack%2Frally~master~I76ead5b94ac2f5f0e3593613b75114567cb254ac","project":"openstack/rally","branch":"master","topic":"ci-jobs","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-07-10 13:04:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I76ead5b94ac2f5f0e3593613b75114567cb254ac","subject":"Add couple rally-openstack jobs to check compatibility","status":"MERGED","created":"2026-07-09 19:03:23.000000000","updated":"2026-07-10 14:14:47.000000000","submitted":"2026-07-10 14:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996706-ci-jobs","meta_rev_id":"058a782a412b557f26e65fdec574784298e93149","_number":996706,"virtual_id_number":996706,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-10 14:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-07-10 13:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-07-10 13:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-09 19:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f987179799f6289b683f1bd076e81d38494d25db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-09 19:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4926e52e03c3b9d92663ce115fa60a9b57904f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-09 19:11:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1bd2fc4596c5bad1ba1f72de2d023924f01c6b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-09 19:12:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a3cc79f043186b45f2dc637537fed332ff88ba12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-09 19:23:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"742bada742ec4fb1e16901a8d4d3e548f0131e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 19:38:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a485918e3f8494bb7d10cb62561bf34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c459fcc6c3e4c50ad0aa3b3aa40a6f0 : SUCCESS in 3m 42s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/4554a5d500514659854e8fd7157b62c8 : SUCCESS in 3m 14s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/fcc05537657042dd85ac46fa1b19dd67 : SUCCESS in 2m 29s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/69983ecf573d44baa13bf207ba9b0e22 : SUCCESS in 3m 07s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/44bfa8be6a444aa394671f66cd0e5857 : SUCCESS in 2m 47s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b4f9ef25d3eb4d45a92d0c2b4fdbeaa7 : SUCCESS in 3m 01s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e6c5d44406414013b9c42c5f165a2943 : SUCCESS in 2m 40s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/57f112a7cd3843b08bdb9ea406269718 : SUCCESS in 2m 23s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/1f3a09e67ccc455ca6c711ea5bd6ddf2 : SUCCESS in 2m 48s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/c0fef1c2e4db453a8e38e4a98fe0996c : SUCCESS in 3m 47s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e1a3b62c1875451aa81b82e6b5daeb8e : SUCCESS in 5m 49s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/70b5384e4caa4a5c81808b95434813ae : SUCCESS in 5m 34s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/557d44d658234150a56148c6ae924dde : SUCCESS in 6m 08s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90efe88f7fd34cc2bb45d726b1b41fe6 : SUCCESS in 5m 06s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/6cceea12358148d9aa9f1da8b80b9a1f : SUCCESS in 8m 25s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/f6d9b0759d4d4220be322f1fb371a1b4 : FAILURE in 12m 00s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/f683cd641bff445ca539e307f8213f8c : POST_FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":4},{"id":"dae8dcaa19a1b771fa7d7ccfd1f75e03ca133667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-09 19:51:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"51cac977967dbe29d7202f9fdae007f7558f990a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-09 20:24:00.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c828509013f14f35905afadf5f67b7139e0d3413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 21:15:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/065f08a231af4ca3a4313bc6831209b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02c61ae1d044de4848e98f57729543d : SUCCESS in 4m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/97069573dff647d0a6d1140b5ce0ba48 : SUCCESS in 4m 21s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0b524cc3e3da4e6597d8375fd014b3fc : SUCCESS in 3m 38s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/fa2e158a20b64a5e9de2ed0e9b25a91e : SUCCESS in 2m 52s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1f65257677481f923ab0d2a2c3f961 : SUCCESS in 1m 46s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/caafff62ffb9457093ed16ff52c94d25 : SUCCESS in 2m 15s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/66902fb12bc241c99d21918fffb2ece0 : SUCCESS in 2m 39s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/fbfce6f0133d4d27ae7302ddacc83a29 : SUCCESS in 3m 09s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/50d7ef559c4e42e393ff6a3bcb36acf6 : SUCCESS in 2m 13s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/0c8fadfa7f9443f1ab29581fead43cfc : SUCCESS in 2m 44s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/58b9749a0517434fb8090dc457dea7bb : SUCCESS in 6m 53s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c621d3d1d03441d5af74ee7ac524c901 : SUCCESS in 6m 30s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7336fae7d87846dfab46db216ea2bb2f : SUCCESS in 5m 33s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b195128d01f745c9be384b0f741623a7 : SUCCESS in 6m 17s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/74c1baf7db1b4fbba1ccb66704fb13b2 : SUCCESS in 8m 09s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/75de5e2e9f7248af9d076713db0478a1 : FAILURE in 25m 37s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/9e65c838925942f09315ceddd9a6e8ba : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":6},{"id":"ce05744d8745c1d7bb0d909c36042f98c656c52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-10 07:21:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"39631dd9009ba67b58d2fd86fbe7c11033750e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 07:40:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59a28cc09908483a95408eb93a02215d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4d84ba85df4a72b447da800f43e4e7 : SUCCESS in 4m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/5e5c15703ecd43bc8b56bcf9528ba823 : SUCCESS in 5m 28s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/283091fc84b040d7a33e4a71c903e24c : SUCCESS in 2m 26s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/713600f1615241de823da606a7bf74a9 : SUCCESS in 2m 04s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce173e5f93f4effabfc9a329dd70d95 : SUCCESS in 2m 21s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/83afac2659de4e15bd28780abecd1ba4 : SUCCESS in 2m 06s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/48d29c1ef8814b1c98cfdcb2bd7e2982 : SUCCESS in 2m 20s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/f47ae227211e41bc9cc82c2bb43a4c30 : SUCCESS in 2m 06s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/c4917f703cdd46d2bd7b215df8822bc3 : SUCCESS in 2m 26s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/938eb5d428f2437784c18efb32aab1a5 : SUCCESS in 1m 48s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7f830234996945cdb31c88229b92764a : SUCCESS in 7m 23s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/30ca489d8cb944d1bd97566945ea4767 : SUCCESS in 5m 45s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0557cd90528a4fb1a99f583d7db78476 : SUCCESS in 5m 02s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42b7ccf0904c4823a65066dbb069b9ad : SUCCESS in 6m 10s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/324d6cc9e0b5426b85bfe663f5d72538 : SUCCESS in 8m 31s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/f821fdc73f59458d8ac1f6785ab5cdad : FAILURE in 14m 41s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/c83feed8efad4e9c8abb38329da53a4a : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":7},{"id":"71221d01556345e5d775c189898cd881030d3c12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-10 08:14:59.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a84254860379d9e0d4c2df22fe9055ddcd10c12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 08:58:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9dd61eb10840d9b088ffa0142cb171\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a424865a382438ca91b6f4ddfdddf02 : SUCCESS in 4m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bfd8bd7f8a774069a53d6a28009c5365 : SUCCESS in 4m 54s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7fc8e0832e1c4166a2ec70e6bd2e402b : SUCCESS in 5m 02s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/349c6b8df9f740a483ad93a5e7fcc1ae : SUCCESS in 3m 27s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bee98bdd7145f5b089b003bc8e275a : SUCCESS in 2m 42s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/2fbea12dd5964586a4ddb9366cbf887b : SUCCESS in 4m 10s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/c4f2586656ad48759069b943fbeb6a6e : SUCCESS in 3m 54s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/b8198042f22a48a380a28e79b88f055d : SUCCESS in 2m 43s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/c358876492c841f7a248d1b19aa82242 : SUCCESS in 3m 14s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8270b290d8da4d1e8b9366d1748a00db : SUCCESS in 2m 55s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6b56ed3407b940259a4ac115be844425 : SUCCESS in 5m 59s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/de0f68d2ee6a41a6a0ab9b44d82b806b : SUCCESS in 6m 00s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33117e0cd2c04c4d9678fe5628f7e9cb : SUCCESS in 6m 14s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/699a308f832b4036b5d24f7de08a791f : SUCCESS in 6m 27s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bfe2d144f0d64c84a27dd6626d04b31d : SUCCESS in 7m 08s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/15a0a14ff13d46659e4e6c3a4caafb7d : SUCCESS in 29m 34s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/b3bc8a1319b84aa194976894ab511c64 : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":8},{"id":"6e391f1d820b736f42fd235ffbf2a25fd0e6911e","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-10 13:04:13.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"92416cd3702ce6d1274bbcdcc01ccb7ffebfc50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:07:36.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5095d8eb34bfa663dc65427097a6c82a5a35b82c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 14:14:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28281810ae3243f284c9276cf4243e93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f7af9f8e7f4a7e9aa6d129c1cdad15 : SUCCESS in 4m 29s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0c290a5f8f7f4cb3999f978624bf28c5 : SUCCESS in 5m 55s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/c8b410e2e1e84c5d823aef29653e9a06 : SUCCESS in 4m 11s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a7fabe8554f24e22ab650a58980fbd66 : SUCCESS in 2m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f1a4c875ef4e1c95c2f03c6e8bab55 : SUCCESS in 2m 50s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/7985c703c270456086ff2e8ed9c30f01 : SUCCESS in 3m 23s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/38b9a5e71d454f279c6a35b15ca6f1ed : SUCCESS in 1m 56s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/2b3d04ce6c9646eca0f975edfdd541cc : SUCCESS in 3m 11s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/8797a41687ff47f1b77569c2ed0d38c2 : SUCCESS in 2m 47s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ad2bf21f011b43ff846c1a7645c77e07 : SUCCESS in 2m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/966c6f06363e45849d8548b79c0b32a8 : SUCCESS in 6m 45s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/493dc16cc4794d4d9893e2706669cb90 : SUCCESS in 6m 20s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3477d53a4a5d453db66ed596522f162b : SUCCESS in 7m 35s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e832af646eb47fba91e20de5b2d6c3e : SUCCESS in 5m 15s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/97fa45705ed448f28b0120e48f53ce72 : SUCCESS in 8m 18s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/93860afb39244bc6a882dcbc883f827f : SUCCESS in 29m 28s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/14101edaa6ec4507bf298f5fe273a9c3 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":8},{"id":"058a782a412b557f26e65fdec574784298e93149","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 14:14:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4a07817be821cdacf74e3de0561cfcc5c6352bfd","revisions":{"47c675fba9ad1aa89df09d067dec303955a54c10":{"kind":"REWORK","_number":1,"created":"2026-07-09 19:03:23.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/1"}}},"commit":{"parents":[{"commit":"e1c55629e2df0761a96cb3918c5f8e58195c16a9","subject":"Update upper-constraints file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e1c55629e2df0761a96cb3918c5f8e58195c16a9"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:43.000000000","tz":120},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/47c675fba9ad1aa89df09d067dec303955a54c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/47c675fba9ad1aa89df09d067dec303955a54c10"}]},"branch":"refs/heads/master"},"1b88eac36e5dcdcaffdcc6d4e45913102f685352":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-09 19:11:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/2"}}},"commit":{"parents":[{"commit":"1b172eefd6cb2a4f34126493fa1a447e4a827c72","subject":"Update upper-constraints file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1b172eefd6cb2a4f34126493fa1a447e4a827c72"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:11:35.000000000","tz":120},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1b88eac36e5dcdcaffdcc6d4e45913102f685352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1b88eac36e5dcdcaffdcc6d4e45913102f685352"}]},"branch":"refs/heads/master"},"4d0015e85a84c1c650547ceac835f316061f2b66":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-09 19:12:40.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/3"}}},"commit":{"parents":[{"commit":"6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3","subject":"Submit a spec about cli modernization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:12:40.000000000","tz":0},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4d0015e85a84c1c650547ceac835f316061f2b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4d0015e85a84c1c650547ceac835f316061f2b66"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1b172eefd6cb2a4f34126493fa1a447e4a827c72","ours":"1b88eac36e5dcdcaffdcc6d4e45913102f685352","theirs":"6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3","merge_strategy":"recursive","contains_conflicts":false}},"e6487e09671020bd7c53c80a231082ab04a7aa57":{"kind":"REWORK","_number":4,"created":"2026-07-09 19:23:01.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/4"}}},"commit":{"parents":[{"commit":"1b172eefd6cb2a4f34126493fa1a447e4a827c72","subject":"Update upper-constraints file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1b172eefd6cb2a4f34126493fa1a447e4a827c72"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:22:48.000000000","tz":120},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e6487e09671020bd7c53c80a231082ab04a7aa57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e6487e09671020bd7c53c80a231082ab04a7aa57"}]},"branch":"refs/heads/master"},"b817be57eb4cad4ecfa4d4101c0ddf3784bd04d9":{"kind":"REWORK","_number":5,"created":"2026-07-09 19:51:11.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/5"}}},"commit":{"parents":[{"commit":"6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3","subject":"Submit a spec about cli modernization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:50:57.000000000","tz":120},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nDepends-On: I9ad68f21c28c94c145181405156d0a411f46d542\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b817be57eb4cad4ecfa4d4101c0ddf3784bd04d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b817be57eb4cad4ecfa4d4101c0ddf3784bd04d9"}]},"branch":"refs/heads/master"},"71f965120798c75b1baf733e3769b3804db549da":{"kind":"REWORK","_number":6,"created":"2026-07-09 20:24:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/6"}}},"commit":{"parents":[{"commit":"6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3","subject":"Submit a spec about cli modernization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6bb4f155320e0fd8709eca1cdfdeda0919d6d1b3"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 20:21:10.000000000","tz":0},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nDepends-On: I9ad68f21c28c94c145181405156d0a411f46d542\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/71f965120798c75b1baf733e3769b3804db549da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/71f965120798c75b1baf733e3769b3804db549da"}]},"branch":"refs/heads/master"},"99141a49421322835f6c7498e76c6bf46b44017f":{"kind":"REWORK","_number":7,"created":"2026-07-10 07:21:51.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/7"}}},"commit":{"parents":[{"commit":"138c3446d1b15e9fec62ee30a3f26eaf17b3ed6f","subject":"Update upper-constraints file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/138c3446d1b15e9fec62ee30a3f26eaf17b3ed6f"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-10 07:21:40.000000000","tz":120},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nDepends-On: I7590427d95d0844cda6eb38a8025d95d15d69bf7\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/99141a49421322835f6c7498e76c6bf46b44017f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/99141a49421322835f6c7498e76c6bf46b44017f"}]},"branch":"refs/heads/master"},"4a07817be821cdacf74e3de0561cfcc5c6352bfd":{"kind":"REWORK","_number":8,"created":"2026-07-10 08:14:59.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/996706/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/06/996706/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/06/996706/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/06/996706/8"}}},"commit":{"parents":[{"commit":"138c3446d1b15e9fec62ee30a3f26eaf17b3ed6f","subject":"Update upper-constraints file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/138c3446d1b15e9fec62ee30a3f26eaf17b3ed6f"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-09 19:02:32.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-10 08:14:48.000000000","tz":120},"subject":"Add couple rally-openstack jobs to check compatibility","message":"Add couple rally-openstack jobs to check compatibility\n\nDepends-On: I7590427d95d0844cda6eb38a8025d95d15d69bf7\nChange-Id: I76ead5b94ac2f5f0e3593613b75114567cb254ac\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a07817be821cdacf74e3de0561cfcc5c6352bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4a07817be821cdacf74e3de0561cfcc5c6352bfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
