)]}'
{"id":"openstack%2Frally~998279","triplet_id":"openstack%2Frally~master~Icd7be1d4c55fae4c69c9cc6a91986e852787b1bf","project":"openstack/rally","branch":"master","topic":"try-trimmed-docker-creds","hashtags":[],"change_id":"Icd7be1d4c55fae4c69c9cc6a91986e852787b1bf","subject":"[ci] Try fixing docker jwt auth call","status":"MERGED","created":"2026-07-22 08:19:06.000000000","updated":"2026-07-22 10:05:22.000000000","submitted":"2026-07-22 09:44:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998279-try-trimmed-docker-creds","meta_rev_id":"33faaabf0610018a792198208f85038b0c498857","_number":998279,"virtual_id_number":998279,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 09:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-07-22 08:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-07-22 08:19:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 09:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3add0352f5ac3766ce23cd97c5200fa9ef9265f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-22 08:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb1c0953f9ee7aec6cbb74119c9da7aa17e0c2b0","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-22 08:19:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b2386381b6008828afb5890ea43eb9df869a9e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:04:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a6a641d52184225b7d7c6fc5cddb69d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665093f628b5494887836fe8df681730 : SUCCESS in 4m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/311618a11c5d4b42973cf9da18b05fb3 : SUCCESS in 4m 55s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e0192b477bf64d9ab2347496ed0ca132 : SUCCESS in 4m 19s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/f67bedd2d4054766ac5db557d1d23d03 : SUCCESS in 3m 22s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e5022a892e42ae84ad8cac52e39b1a : SUCCESS in 2m 30s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/74393727fc0a423c81b9e398658cae70 : SUCCESS in 3m 14s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/c90db6f8184946fe93f0b11f1e982e3d : SUCCESS in 2m 56s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/3b716bbc9aab420c91e521f2e10a7f69 : SUCCESS in 4m 22s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/a5ee3cb157044465b466aff8bd1a7dd0 : SUCCESS in 3m 47s\n- rally-tox-zuul-ansible-lint https://zuul.opendev.org/t/openstack/build/0853a4dd821d4ce7bf0e0126ab5246a4 : SUCCESS in 4m 12s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e3e93e1e855747a8ba38c1396baecf5e : SUCCESS in 3m 09s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4dec497b839e4feb8df0b0925b46aeae : SUCCESS in 5m 41s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/bd054068bba7498ca98229f6b096c987 : SUCCESS in 7m 15s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1726182f93a8473a89780a4dd24a10d5 : SUCCESS in 7m 17s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c92facc04224ffe87724c47e3e0594a : SUCCESS in 8m 36s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2f1b151f1aa540f4a1e707183422ddc0 : SUCCESS in 8m 26s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/7e10323193254569ac08e2b7bd139e8e : SUCCESS in 28m 57s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/f2cd69b194944b14804e92df89cf156d : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":1},{"id":"cfc4f34c83d03330d729599e9010728087b59388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:05:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38c91dacd225f069bff18fafee27771eb4c60607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:44:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83977c603eef4ad7b50e97f909a4f7b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d68393129a4432cb7f8f24e969dcd91 : SUCCESS in 4m 31s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bf74e2a264ff4ad6a40ba506b2b950b2 : SUCCESS in 4m 49s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/732e57645ef94c1c90ca2d5843a40fa4 : SUCCESS in 4m 02s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/07fdbfcb73444f6792365c53ecc1d80e : SUCCESS in 3m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/492f5a013fd74302802cfda1f670e4bb : SUCCESS in 2m 39s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/38bcd0882ec24b43aafd4896612c68b7 : SUCCESS in 3m 16s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/8ffe809c42044c40817377422be24e54 : SUCCESS in 3m 41s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/8bc02ae068f74b74a8edd000c9a0ec87 : SUCCESS in 3m 11s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/23ce1e654234453798628a79c9609d6e : SUCCESS in 4m 28s\n- rally-tox-zuul-ansible-lint https://zuul.opendev.org/t/openstack/build/75dd0b93b9af4c9da54b03d5dc91fc92 : SUCCESS in 2m 16s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/c48f36b7e48649ac8ce6b69124b35125 : SUCCESS in 2m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1a5f2339272b4b1e84d8a9a869d5c86b : SUCCESS in 6m 01s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ee2485853af14196800ef2c33f2abca2 : SUCCESS in 5m 20s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d14d9917900a4b8199953a7a4d0fb2c6 : SUCCESS in 7m 14s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2594f16404614d34a7d0804fdcbb5f2e : SUCCESS in 5m 04s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/51085ada127a423f899b42b861f73929 : SUCCESS in 8m 09s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/bf3d8f75198042d28da6c39f2cf76abc : SUCCESS in 28m 05s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/3e66a5eb47cb4b4fa24ee8083b0389eb : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b0ba26369785147bed128bc37ac03e501fa267cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:44:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"33faaabf0610018a792198208f85038b0c498857","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-22 10:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9dbd846ae9619a9494127baba915cbcb0309b7e6","revisions":{"9dbd846ae9619a9494127baba915cbcb0309b7e6":{"kind":"REWORK","_number":1,"created":"2026-07-22 08:19:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/79/998279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/79/998279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/998279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/998279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/998279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/79/998279/1"}}},"commit":{"parents":[{"commit":"9d33a067e3ab7c61b9bd3ac24109c75c68af3d33","subject":"[ci] print more details about docker push steps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9d33a067e3ab7c61b9bd3ac24109c75c68af3d33"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-22 08:18:34.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-22 08:18:59.000000000","tz":120},"subject":"[ci] Try fixing docker jwt auth call","message":"[ci] Try fixing docker jwt auth call\n\nChange-Id: Icd7be1d4c55fae4c69c9cc6a91986e852787b1bf\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9dbd846ae9619a9494127baba915cbcb0309b7e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9dbd846ae9619a9494127baba915cbcb0309b7e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
