)]}'
{"id":"openstack%2Frally~999294","triplet_id":"openstack%2Frally~master~I9e3e70f001f89d84b9e8fe585552ee3caba33a9e","project":"openstack/rally","branch":"master","topic":"type-adapter","hashtags":[],"change_id":"I9e3e70f001f89d84b9e8fe585552ee3caba33a9e","subject":"Provide a helper to map an arg spec to signature of function","status":"MERGED","created":"2026-07-30 14:00:40.000000000","updated":"2026-07-30 15:18:59.000000000","submitted":"2026-07-30 15:18:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999294-type-adapter","meta_rev_id":"12173dde08d339be45ebecde59c7a087b9e38ded","_number":999294,"virtual_id_number":999294,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-30 15:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-07-30 14:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-07-30 14:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-30 14:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c96290fbad06a09d5567effcd4ef0f553864869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-30 14:00:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"066bef1ce41f9a0d8e62608af666082a25a30a19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-30 14:06:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b0b31bac0d850f0543c178eaf2859d4f43c24ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:29:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb56b9531e542a58b91b811a2dabc8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b840e9cc1a2f41e1ab546acf02747dfe : SUCCESS in 3m 17s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2137eda310534516b598fb75901d9875 : SUCCESS in 2m 56s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9b9e9862e5bf4cd5a0e11f56955e3aab : SUCCESS in 4m 07s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a08215a214ec4deba5769992c943fd37 : SUCCESS in 2m 52s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/651a310c22dd439c9134dc727cc0144d : SUCCESS in 2m 11s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/cf0130f2c79849b397a2e88e4541775b : SUCCESS in 3m 32s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/6d8338fe6db144ce93835d30afe208df : SUCCESS in 2m 27s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/e7ec62eee2244b56a0e0a058bcf92f29 : SUCCESS in 2m 06s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/078ba6789cd14526b555fe1a2731d961 : SUCCESS in 2m 09s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/8df73c16fd684f2aa215bc2e35fa451d : SUCCESS in 2m 21s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c19da6d5d09448938e1ee2c65a3d5806 : SUCCESS in 5m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/ca8a6774b5894403adbcfed8272d5ecb : SUCCESS in 5m 49s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23fea1fa0a18449fadcfc6713391d68e : SUCCESS in 6m 20s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67b83ab8de4041fa9bf5bfed7e98acc9 : SUCCESS in 5m 33s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/59c319ad93aa4fe5b115751d2713b077 : SUCCESS in 6m 33s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/cdd5e45c62c1452c8a3c77c6ef2c993e : SUCCESS in 13m 23s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/fc06d15b81934d32bfe03cf993cb23a4 : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"17b73c1c5ef7769658aee69f382a2c72323916e8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-30 14:36:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"afc7b932be3b96d8ceca4c6a339d6fdf3b3fc796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:37:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da463124ce2014d71b1a7730de140bfb2071696f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 15:18:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbfc54d3cf9479cafbaab09275fecc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221275e997f840ad90ac64cfa4cc1c17 : SUCCESS in 4m 36s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bfb49e15d3c94a18b9310ff1694616fd : SUCCESS in 4m 43s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a51895a3e64047d28c3812425bafae1f : SUCCESS in 2m 44s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a3c306bd92ef4979a0d6bbc077b65ab1 : SUCCESS in 3m 51s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/74fbcacdffca45ffab260bbad0c29271 : SUCCESS in 2m 32s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/5c6abbbf5bec41c19ee9973a043d73c2 : SUCCESS in 2m 31s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/738096b139d6444a9bca180db273aa61 : SUCCESS in 3m 19s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/0779d328d58349f182915bb1b9d76386 : SUCCESS in 3m 27s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/9b2d3874b98141d1b26ff92a40566b12 : SUCCESS in 3m 02s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/955cbd429560458182defb9738975db9 : SUCCESS in 3m 24s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3f68aba964df42fb844c4aa343f7d28c : SUCCESS in 6m 14s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/3d7194b77607453cbb09c60682f7c150 : SUCCESS in 7m 48s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9015858befa843f5b67858728bfd865e : SUCCESS in 6m 22s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b55a90635584ef79230edc16bb5b63f : SUCCESS in 5m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/d62afdbdee494c1cb9f024f356fbb07d : SUCCESS in 8m 20s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/bcc5e9f4e00e4299b3c295d164b86977 : SUCCESS in 30m 45s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/7ce60b82e3ad4e01b85cc0b181358584 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":2},{"id":"12173dde08d339be45ebecde59c7a087b9e38ded","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 15:18:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"219370b4cb97af19e4f1e9da95c1bbcd8312e34d","revisions":{"d0e84dfa515532e34202321b83631268a1370956":{"kind":"REWORK","_number":1,"created":"2026-07-30 14:00:40.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/94/999294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/94/999294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/999294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/999294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/999294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/94/999294/1"}}},"commit":{"parents":[{"commit":"9dbd846ae9619a9494127baba915cbcb0309b7e6","subject":"[ci] Try fixing docker jwt auth call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9dbd846ae9619a9494127baba915cbcb0309b7e6"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-30 13:59:58.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-30 13:59:58.000000000","tz":120},"subject":"Provide a helper to map an arg spec to signature of function","message":"Provide a helper to map an arg spec to signature of function\n\nChange-Id: I9e3e70f001f89d84b9e8fe585552ee3caba33a9e\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d0e84dfa515532e34202321b83631268a1370956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d0e84dfa515532e34202321b83631268a1370956"}]},"branch":"refs/heads/master"},"219370b4cb97af19e4f1e9da95c1bbcd8312e34d":{"kind":"REWORK","_number":2,"created":"2026-07-30 14:06:42.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/94/999294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/94/999294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/999294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/999294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/94/999294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/94/999294/2"}}},"commit":{"parents":[{"commit":"9dbd846ae9619a9494127baba915cbcb0309b7e6","subject":"[ci] Try fixing docker jwt auth call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9dbd846ae9619a9494127baba915cbcb0309b7e6"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-30 13:59:58.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-07-30 14:06:21.000000000","tz":120},"subject":"Provide a helper to map an arg spec to signature of function","message":"Provide a helper to map an arg spec to signature of function\n\nChange-Id: I9e3e70f001f89d84b9e8fe585552ee3caba33a9e\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/219370b4cb97af19e4f1e9da95c1bbcd8312e34d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/219370b4cb97af19e4f1e9da95c1bbcd8312e34d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
