)]}'
{"id":"openstack%2Frbd-iscsi-client~852783","triplet_id":"openstack%2Frbd-iscsi-client~master~I0042e16d19e133546b0384fec5a6aee1701eda28","project":"openstack/rbd-iscsi-client","branch":"master","topic":"sync-u-c","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-09-09 14:31:38.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-09-09 14:31:38.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-09-09 14:31:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0042e16d19e133546b0384fec5a6aee1701eda28","subject":"Replace testing template with individual jobs","status":"MERGED","created":"2022-08-10 17:56:55.000000000","updated":"2022-09-09 14:34:00.000000000","submitted":"2022-09-09 14:31:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"852783-sync-u-c","meta_rev_id":"a16375f3c37865ac92110e8680644194d40f089b","_number":852783,"virtual_id_number":852783,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 14:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-09 14:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-08-17 18:53:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-09 14:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-17 14:40:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 18:01:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 07:38:52.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-08-17 08:10:15.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-09-09 14:25:14.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"8604c586ea50e6d2c2b49b049a72beb56a7bf2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-10 17:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9cd5e2fd326b6f826fece2989aac57da57a65e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:01:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/899b0a65e416460191e99cd5bb8dae8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae3a441eaa94ec287681ca124ff65ad : FAILURE in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09d0df303d8840df8ff9428882dd235b : FAILURE in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98fef326c088469994926f162260f562 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62991cb6de8d42ccb3bc5ceabbed8d82 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afda7893a604354a04a3735f31dba6f : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a807f0ecab3485bafcbb821ee629cf6 : SUCCESS in 3m 14s\n\nWarning:\n  Comments left for invalid file rbd_iscsi_client/client.py","accounts_in_message":[],"_revision_number":1},{"id":"37f34fd7120521784b35ef00afefd4857522b83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-10 18:39:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af8fdcbfa38cc1b80ad6ce5e53bce3e47d44f600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:44:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65befe1a783c408eafecb946cbb9aba0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576055702f9f40dba1c73e25521d1209 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa04ddaf5a4e4a769eb5c31280a9c4ed : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15eefd685f49436cba42134828d9925b : FAILURE in 3m 09s (non-voting)\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/474d81f4a66040e891b7c5aaaa5c6359 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f5bf5ec8c0144e9b298f5b17afd8421 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09e23c7924f44be2bac3dd443ab5c57b : SUCCESS in 3m 59s\n- rbd-iscsi-client-code-coverage https://zuul.opendev.org/t/openstack/build/a6f921c42a7a477d8a62001746af1a89 : SUCCESS in 3m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file rbd_iscsi_client/client.py","accounts_in_message":[],"_revision_number":2},{"id":"61a106997cabaec9db86a411ac6584fb2f83fbcb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-10 20:43:41.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"83a5a3a731b8700d63c9d82cdea64bedc343fa65","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-17 07:38:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7115c2fc9d4893a11de8bd8f5323fa635cb98ae3","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-17 08:10:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"45b8d8adfee171dec52d226917dc01c64e5aa624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-17 14:31:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4091a9b1400bc5ddf5bf1edb5d6b452240a585ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 14:40:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbdbca1630e341f29d871de0beee04c7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb067879c62a4de3a08b8daf1af7d433 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade10346cfc143bc81f655f9c4914303 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1643a95efd584254b39b471bddb723a6 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfb2b5f1d9e46de9f419c195c54d476 : FAILURE in 3m 42s (non-voting)\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/3be4e180678e4ace9f0380b833697643 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12af560ddc5647138f9900ec866b6fa1 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d6a7f5270143119320c9a742fa98bd : SUCCESS in 3m 52s\n- rbd-iscsi-client-code-coverage https://zuul.opendev.org/t/openstack/build/8837e9a64b9e4267a1701880222bed02 : SUCCESS in 6m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file rbd_iscsi_client/client.py","accounts_in_message":[],"_revision_number":3},{"id":"b43621ef43f5c02333c4b940bdf0d71784010818","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-17 14:40:20.000000000","message":"Patch Set 3: Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"051dbd43fcfa9f3ec3d2ab1e3fdcbded9407b9b0","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-17 18:53:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6419f55af7d1eb56bbb448490b8e0f2fb2792570","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-18 11:55:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb1c0a9157dcee61dd0f12078df26d0a13d612ab","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-09-09 14:25:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db4ca3f02aba67221b4589fef5cb4d921dd1b7fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 14:25:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ead3cdb86381b274c29222b3b6c449ca162a43b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 14:31:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ce49b4437646ad931981078f1a3060\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5c1af18c2034df1b4001beb58ffa535 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf70b9b51ae404db11ecdfd714f3944 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84eef2087dcb46c79d63a753845d360e : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3dafb4f09024758ae9a55b95f3fcad2 : FAILURE in 3m 45s (non-voting)\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/c806a8edefd04ada8db39d1631db510f : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dadd9dba60446a59dbab6a6de93baf8 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4385e90d94f644c9b6aba9b5fa7e52b4 : SUCCESS in 3m 56s\n\nWarning:\n  Comments left for invalid file rbd_iscsi_client/client.py","accounts_in_message":[],"_revision_number":3},{"id":"0d263aefa81f53f90f125cd0d1d83fbf7a193ee3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 14:31:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a16375f3c37865ac92110e8680644194d40f089b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 14:34:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe9f41d04764ad5ae320390eaf875cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e646a416d5f437a92f57a080ce2efc3 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa7e8f4e37794093b34d03703858b631 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe63f0cc892a1598654ec09f086512915d1d154e","revisions":{"ac5193c46a41dd9f287720518827f8e15ac0b9f0":{"kind":"REWORK","_number":1,"created":"2022-08-10 17:56:55.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/83/852783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/83/852783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/1"}}},"commit":{"parents":[{"commit":"f0227095b9b5cebd60840548bc0de3e81942c02c","subject":"Drop lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f0227095b9b5cebd60840548bc0de3e81942c02c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-10 17:54:59.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-10 17:54:59.000000000","tz":-240},"subject":"Use xena upper constraints","message":"Use xena upper constraints\n\nrbd-iscsi-client still supports py36 and uses the xena testing\ntemplate for CI, so also use the xena upper constraints.\n\nChange-Id: I0042e16d19e133546b0384fec5a6aee1701eda28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/ac5193c46a41dd9f287720518827f8e15ac0b9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/ac5193c46a41dd9f287720518827f8e15ac0b9f0"}]},"branch":"refs/heads/master"},"9d894b2263ab9de91b240330c7f2683a225a0cd8":{"kind":"REWORK","_number":2,"created":"2022-08-10 18:39:28.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/83/852783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/83/852783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/2"}}},"commit":{"parents":[{"commit":"f0227095b9b5cebd60840548bc0de3e81942c02c","subject":"Drop lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f0227095b9b5cebd60840548bc0de3e81942c02c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-10 17:54:59.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-10 18:36:54.000000000","tz":-240},"subject":"Replace testing template with individual jobs","message":"Replace testing template with individual jobs\n\nBeing on an independent release model, rbd-iscsi-client supports\nmore pythons than are supported by current openstack development,\nso adjust the zuul and tox config to reflect this.\n\nAlso temporarily makes the pep8 job non-voting so this change\ncan be merged.  The pep8 job is fixed and restored to voting by\nchange Ia4d40a85ac15.\n\nChange-Id: I0042e16d19e133546b0384fec5a6aee1701eda28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/9d894b2263ab9de91b240330c7f2683a225a0cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/9d894b2263ab9de91b240330c7f2683a225a0cd8"}]},"branch":"refs/heads/master"},"fe63f0cc892a1598654ec09f086512915d1d154e":{"kind":"REWORK","_number":3,"created":"2022-08-17 14:31:23.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/83/852783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/83/852783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/83/852783/3"}}},"commit":{"parents":[{"commit":"f0227095b9b5cebd60840548bc0de3e81942c02c","subject":"Drop lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f0227095b9b5cebd60840548bc0de3e81942c02c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-10 17:54:59.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-17 14:30:42.000000000","tz":-240},"subject":"Replace testing template with individual jobs","message":"Replace testing template with individual jobs\n\nBeing on an independent release model, rbd-iscsi-client supports\nmore pythons than are supported by current openstack development,\nso adjust the zuul and tox config to reflect this.\n\nAlso temporarily makes the pep8 job non-voting so this change\ncan be merged.  The pep8 job is fixed and restored to voting by\nchange Ia4d40a85ac15.\n\nChange-Id: I0042e16d19e133546b0384fec5a6aee1701eda28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/fe63f0cc892a1598654ec09f086512915d1d154e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/fe63f0cc892a1598654ec09f086512915d1d154e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[]}
