)]}'
{"id":"openstack%2Frbd-iscsi-client~868324","triplet_id":"openstack%2Frbd-iscsi-client~master~I651f762bf6aac9679d1fd74817e7e5f201ced774","project":"openstack/rbd-iscsi-client","branch":"master","topic":"tox-4-ready","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-12-23 16:27:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I651f762bf6aac9679d1fd74817e7e5f201ced774","subject":"Get ready for tox 4","status":"MERGED","created":"2022-12-21 14:21:05.000000000","updated":"2022-12-23 16:29:16.000000000","submitted":"2022-12-23 16:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868324-tox-4-ready","meta_rev_id":"c7bbede92b1279f1a28ab1bb6ba9b821ac88c85a","_number":868324,"virtual_id_number":868324,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-23 16:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-23 16:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-23 16:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 14:35:42.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-12-21 14:37:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-21 14:45:13.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2022-12-23 16:16:09.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"09b21037a835bf55496e836c88ba2e03814eaa53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-21 14:21:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0972037d97141989260340e3885ca3238124eb02","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-12-21 14:35:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7e65bfd1cd1d314723f7fd7ac3683de93307ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 14:37:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/133fa4c79f2f44beb9e5da0ca3dc83d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198d7a1ead674cefb6551e20e94f2cf2 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ce0d68c73249879bf23d2258386e99 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/770c03bd0f5640fe9c7701025ace8fb8 : SUCCESS in 2m 52s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/2c7a99b970b64ca0aa035e4a21dd3f07 : FAILURE in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60b12ca8102c4918bbc95b1311582e1d : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b0ff46df08048a188ad437f0b6563b0 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"515d7139eca25fe261a6c542d28f3ae02905507d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-21 14:45:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85aea2bb5c75dbe4b6094bc59fdb36d05b5882b1","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-21 14:45:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d49a281573775c7fb254ce65c2db045ddd8f6493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 14:53:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/705a5a45e3dc4a478cb51d951ff0e413\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6a7be509374eafa067057b111ca559 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f2fcbdea2af45bea7c7011b52ae7dbb : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a3e9226faa4e6d8586eafeab9d59e4 : SUCCESS in 2m 40s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/53acb83e71bf4b0a91b923a328c45ed4 : FAILURE in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec064871bb5741328288ab767e4f2f82 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05d500561b57447daebfb218f547147b : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7dcd34842560b6315b5c9951a3bf59bbbbf112b1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-21 15:21:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b088a065180099511c694e0084d3c7bc562123de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-21 20:58:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1bbe59d1d896efff29bddce6133e347fa0752651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 21:06:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67788949877a454fbdfb78c809e74122\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49248130baa42199c9cbf31926e7b87 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31646857f8c54e1e8f8b8d250cfefbb6 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71ca0f5fc4f4f71a348447fd12017a5 : SUCCESS in 2m 26s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/5b411abc34da4739ba49df9cfed69d62 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9a809dcb8214dfe88f19099b5a95d5f : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e3a9d2e3f3040669c4cdab7791105aa : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"876d61230b2a57a29d557849194f9612a04161be","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-23 16:16:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e70aa71c9a73505b123915bde5411bc1fe682606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:16:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cee2aa3cb3ac7aad8cca117ecd3a63de05046632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:27:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a381500fed44c8094686ff490727393\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6160eddfa8d04ae7a90f02bcec590e6f : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc40ce45d82040caba71d882e0fd3b35 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0906bbe4c8e43c7a14556259d5b65e6 : SUCCESS in 2m 18s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/2cc8ce448bed46f48e444ef95cfff016 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef6fa61605d848d080a0031cf73a6c19 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad077b1debf47fd9e4e398489fd54a8 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"557c346c08f7424779b1a6bf1a7f62aeb6dc7df2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:27:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c7bbede92b1279f1a28ab1bb6ba9b821ac88c85a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:29:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78dceb8b1f37435b9a99df78628b5ae9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aff73e2c8e746aaaa8ebc7bea38badd : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0f5178c18fc4bedae6004178624d629 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb","revisions":{"a3f6f49ddc21aaee9b5a10a0940bf6d5f50d18ca":{"kind":"REWORK","_number":1,"created":"2022-12-21 14:21:05.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/24/868324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/24/868324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/1"}}},"commit":{"parents":[{"commit":"8fe7c8a5d3c7d3233ed12d65cfc9bbf1b9f174f7","subject":"Address G002 in flake8-logging-format 0.7.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/8fe7c8a5d3c7d3233ed12d65cfc9bbf1b9f174f7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-21 14:18:06.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-21 14:20:49.000000000","tz":-300},"subject":"Get ready for tox 4","message":"Get ready for tox 4\n\nChanges:\n- eliminate whitespace in passenv values\n- account for stricter allowlist checking\n- removed skipsdist\u003dTrue, which in tox 4 appears to prevent\n  rbd-iscsi-client from being installed in the testenvs\n- made 4.0.0 the tox minversion\n- updated the \u0027bindep\u0027 testenv so it would work if we had\n  a bindep.txt, which we currently don\u0027t\n\nThis patch makes tox 4 the default so that we can hopefully catch\nproblems locally before they block the gate.\n\nChange-Id: I651f762bf6aac9679d1fd74817e7e5f201ced774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/a3f6f49ddc21aaee9b5a10a0940bf6d5f50d18ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/a3f6f49ddc21aaee9b5a10a0940bf6d5f50d18ca"}]},"branch":"refs/heads/master"},"3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb":{"kind":"REWORK","_number":2,"created":"2022-12-21 20:58:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/24/868324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/24/868324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/24/868324/2"}}},"commit":{"parents":[{"commit":"8fe7c8a5d3c7d3233ed12d65cfc9bbf1b9f174f7","subject":"Address G002 in flake8-logging-format 0.7.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/8fe7c8a5d3c7d3233ed12d65cfc9bbf1b9f174f7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-21 14:18:06.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-21 20:54:28.000000000","tz":-300},"subject":"Get ready for tox 4","message":"Get ready for tox 4\n\nChanges:\n- eliminate whitespace in passenv values\n- account for stricter allowlist checking\n- removed skipsdist\u003dTrue, which in tox 4 appears to prevent\n  rbd-iscsi-client from being installed in the testenvs\n- updated the \u0027bindep\u0027 testenv so it would work if we had\n  a bindep.txt, which we currently don\u0027t\n\nWe can\u0027t make tox 4 the default because tox 4 only supports python\n\u003e\u003d3.7, and rbd-iscsi-client still supports py3.6.  When our CI job\nfor py3.6 lands on a system whose default python is 3.6, the job\nwould fail because it couldn\u0027t install tox 4.\n\nChange-Id: I651f762bf6aac9679d1fd74817e7e5f201ced774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
