)]}'
{"id":"openstack%2Frbd-iscsi-client~880701","triplet_id":"openstack%2Frbd-iscsi-client~master~I7fa0189dac8d980aa4fd5a7c8bac2f6f84ceb6d5","project":"openstack/rbd-iscsi-client","branch":"master","topic":"2023.2-functests","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-04-26 12:53:31.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-04-18 14:31:41.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I7fa0189dac8d980aa4fd5a7c8bac2f6f84ceb6d5","subject":"Support Python 3.10","status":"MERGED","created":"2023-04-18 03:09:57.000000000","updated":"2023-04-26 12:55:19.000000000","submitted":"2023-04-26 12:53:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880701-2023.2-functests","meta_rev_id":"f90b8373cbe78f8cb95ecd4e94074fd945238672","_number":880701,"virtual_id_number":880701,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-26 12:53:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2023-04-18 14:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-26 12:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-26 12:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-18 03:22:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 13:22:44.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-04-26 12:43:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"30871294cb0e2a6d8e5516735e41f513978a59f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-04-18 03:09:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a28fe7327ef5ad2b1b9805fffc6f5f73d2d8f1f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 03:22:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f49b2093837b498d90f4ed0e693964c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3134a54e8ff4d9d92d548ede004e1a3 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be52165ba1545b1ac14cc3f082ba142 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c9a92295e1946b7ab5f37fadb9b381b : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6361ff0e26e94d6a995ec06eb4467f30 : SUCCESS in 2m 45s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/3741d53a3b654d0394c86e6fc229a72d : SUCCESS in 3m 15s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/7e543ed23d224502b2bda7251b94603e : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f37327251f543e99edbf855b89f5690 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de252436acd848459f56155c61147094 : SUCCESS in 3m 22s\n- rbd-iscsi-client-code-coverage https://zuul.opendev.org/t/openstack/build/8c45f1d3bc3a4bd0a43248a90c1ca2b7 : SUCCESS in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9a60fffee7ad8b68daeac0a74f084f5b2f729c9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-18 13:22:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"953c24833a405097d21c269db3a474a639252cd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-04-18 13:35:08.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"286bbacc71738aa6b8aae471bdf9d062a3142704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 13:47:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76026a84b314e41bfc22847daef5c91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c2e88267b01448bbef635b52ac9a9e3 : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14acbf3408b04e2ba745fc86bf52549f : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a686a605bcc94a0c823d08074b90d82c : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d70ec6dafb44ae88003bafc91891e9 : SUCCESS in 2m 16s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/8f5ec4b1d82e4c8583f88fb506971f5f : SUCCESS in 3m 52s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/02e39de54e3f4ffd889d9ae3e8ec6967 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d77a83e6e6d45f2b212fa4ebe46cb2a : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e95de3f51e3d41918fb77d89a30b8778 : SUCCESS in 2m 49s\n- rbd-iscsi-client-code-coverage https://zuul.opendev.org/t/openstack/build/18f467be68d147c8a4bcd8c67d19ab2f : SUCCESS in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ceb1e54ef67d860c20bbf582ba2c0d50be8fe62","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-18 14:31:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bdb811bc69c92ee6da0c11e06506915586d8920d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-18 14:31:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2a068184f4b1e152755eb9e17db6d663d35896d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-04-26 12:43:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22951e6ea196b245833016618bc4878047459da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:44:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85f83c6d4c15098572c1ecf68aae0612a64660c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:53:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da05c09ee35142279deab948ef9ef7eb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56d8b231ea084be48902593ac8eb027b : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ee2276695249a7b293487da525be14 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1694d7a59f6a4e97b6f331a9dce4c518 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded4088e18c643d4b7a10fd6f7b38d19 : SUCCESS in 3m 30s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/b6500ebf7b454ccb9dbd869822247716 : SUCCESS in 5m 12s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/bb1499481de145268d56d517d1f9553a : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e05e852fd154f8bbf09f17da0dae7c2 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec7b7b7e8279425fa8c46851ee46d0c5 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"51ce9a4ce7da239320ca468fcb708fd766abbdd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:53:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f90b8373cbe78f8cb95ecd4e94074fd945238672","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:55:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee095276d89445d0901a70a45c827308\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b710fd34875c431f9664fc55d21308f8 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f14e91daa1da4cd7b5dce5edda854422 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f56f42dcd8908664508bbed53cd56c2728daf4a7","revisions":{"754b0b89ce005360671d40a717d1c8f7c99b00ec":{"kind":"REWORK","_number":1,"created":"2023-04-18 03:09:57.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/01/880701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/01/880701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/1"}}},"commit":{"parents":[{"commit":"3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb","subject":"Get ready for tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-04-18 03:02:59.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-04-18 03:02:59.000000000","tz":-240},"subject":"Support Python 3.10","message":"Support Python 3.10\n\nAdd a CI job to run tests on Python 3.10, which is supported\nin 2023.2 [0].\n\nAs rbd-iscsi-client is on an independent release model, it\nstill supports Python 2.8, so adjust .zuul.yaml and tox.ini\nto use the 2023.1 upper constraints when running py38 tests.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: I7fa0189dac8d980aa4fd5a7c8bac2f6f84ceb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/754b0b89ce005360671d40a717d1c8f7c99b00ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/754b0b89ce005360671d40a717d1c8f7c99b00ec"}]},"branch":"refs/heads/master"},"f56f42dcd8908664508bbed53cd56c2728daf4a7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-18 13:35:08.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/01/880701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/01/880701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/01/880701/2"}}},"commit":{"parents":[{"commit":"3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb","subject":"Get ready for tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/3f7eff31e1c07cbf43913ec92ab2bdffe79f76bb"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-04-18 03:02:59.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-04-18 13:35:08.000000000","tz":0},"subject":"Support Python 3.10","message":"Support Python 3.10\n\nAdd a CI job to run tests on Python 3.10, which is supported\nin 2023.2 [0].\n\nAs rbd-iscsi-client is on an independent release model, it\nstill supports Python 3.8, so adjust .zuul.yaml and tox.ini\nto use the 2023.1 upper constraints when running py38 tests.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: I7fa0189dac8d980aa4fd5a7c8bac2f6f84ceb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f56f42dcd8908664508bbed53cd56c2728daf4a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f56f42dcd8908664508bbed53cd56c2728daf4a7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
