)]}'
{"id":"openstack%2Frbd-iscsi-client~889094","triplet_id":"openstack%2Frbd-iscsi-client~master~Ibbca458dd936f9ec8b93fb8d5d0bc9c1f05f5467","project":"openstack/rbd-iscsi-client","branch":"master","topic":"remove-egg-info","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-12-15 15:45:26.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-12-15 15:45:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbca458dd936f9ec8b93fb8d5d0bc9c1f05f5467","subject":"Remove egg_info from setup.cfg","status":"MERGED","created":"2023-07-20 13:03:30.000000000","updated":"2023-12-15 15:46:19.000000000","submitted":"2023-12-15 15:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889094-remove-egg-info","meta_rev_id":"36fd7526d8636bf3dd140e14d0206c6e90e82ca3","_number":889094,"virtual_id_number":889094,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-15 15:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"all":[{"value":1,"date":"2023-11-21 23:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2023-11-21 23:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":1,"date":"2023-11-21 23:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},{"value":2,"date":"2023-11-21 23:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-15 14:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-15 14:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 13:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 13:01:17.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2023-07-21 13:58:27.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-07-21 14:22:53.000000000","updated_by":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"reviewer":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"state":"REVIEWER"},{"updated":"2023-07-21 14:48:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-11-15 17:12:00.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"2fb07e1b7a69fbc854cb670d3b146a846f6d4e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-20 13:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a9a4ae50e6b5a89663cf5d1b359de022e9a0854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 13:27:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9bb3aceea1d47cd96f2cca607bc25c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b6ddf5e8256451ea4b457979b6a7bd4 : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0559a0713a174b208300ba3da9ff3600 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d888ea9889489da802ae0e00f81215 : SUCCESS in 3m 38s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/b43e45a6f1b648fd8cfe4abf0eda1f15 : SUCCESS in 3m 18s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/3d514203ad67401182f05fa2f9390d00 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18a1bfc8f3e4495db79d8c25438cc460 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e019b9c0d744ab497ac19624e563d56 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a71d5f12f0be10cae80d54b582d40160eb2056d9","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-07-21 13:01:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5614a755e6234d4b2d5d2a727cd98a3b441b6627","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-21 13:58:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"026b4f1f18c9108525865c979a488eef1c555d40","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2023-07-21 14:22:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1ffcc465faa58a3ff369580effcf3a07bc92739","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-21 14:48:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2035c5d3d895492f0d32379e193213f71a702569","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-15 17:12:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa07553aab7b04aef3b82c54a7ff61342cfc83cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:12:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9534a9df2a2f913a7f4628b097625becd5e29f03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:22:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3c1286f6f7f44a4bcf7881911483a89\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/913e0245711e430cbaefac21dda5e612 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf5f30a042e42c8ab784b3ba29791a3 : FAILURE in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e3870f69dd445492d22a39d39562c6 : SUCCESS in 4m 12s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/d16929d91b3b4c2dae6d639e5e8f86a6 : SUCCESS in 2m 43s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/6ab4652cec2a40188201719142275a54 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c098df3381ef41afbc37e82073945a46 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e21a93ebac94812b76d15dbf7b9a3f9 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7a3368d86d93d140bda38b5622df735a0547a1e8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-11-21 23:34:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bf204f3965a6490686e4420aab74e109b9f302d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-11-21 23:35:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"91485ef29a61eb5d1d1f198f21edc16e81187054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 23:46:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/447bd432e9604e79880e38d08a52b1af\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2404efb979714db1807f97e0dd24a455 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f4ab285ecb471bbf19925013e45c7f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca62be5727c4cf295c9c2c5ba606540 : SUCCESS in 2m 44s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/31a61b5bc0a64ae0b3cc97bc4fdf222d : SUCCESS in 3m 02s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/4fc102d8f4c6466daf7a7b2d3aea5f62 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5916afab7154f66bd6981e906476d6e : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3e3735c4cc44396ba68cf863ed3c1d5 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"480fd7e60e08e6a216805e0372080269eedabdd1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-12-15 14:21:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aea929c780ecb7c9242bb5d8d04248e0575ccc4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 15:30:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c7b94cda8bc4e853a302174637db94f3b1a7b49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 15:45:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b42364d260f48d18a61aff1eb0d7121\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82bfcf478b8946a08abfd4290f4493be : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa3bdc7ed1d4fa585b463fb7ddab20d : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c5eef5bdab43c8b47557cbb86a7048 : SUCCESS in 3m 08s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/b3385821bd804709b99743b3fb98ea30 : SUCCESS in 3m 29s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/dcb67637fb574fab8cbf5fd8b9ac2113 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5012177f78414166ada1774f70ae37e1 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a52dcbdb78eb49ec834ecfd81c037aee : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"07c2589f28470244d223008de0f43674e45d75fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 15:45:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"36fd7526d8636bf3dd140e14d0206c6e90e82ca3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 15:46:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41e66378012a48aeb2bf8882946f3a44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5cc555ca79442a84fb5d3055bfd536 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ecfac3da494fc4d593b324e9fb4fe1da52ec913","revisions":{"f5e6b43633b9993a8600660be9c42a3c84e20f5a":{"kind":"REWORK","_number":1,"created":"2023-07-20 13:03:30.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/94/889094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/94/889094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/1"}}},"commit":{"parents":[{"commit":"f56f42dcd8908664508bbed53cd56c2728daf4a7","subject":"Support Python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f56f42dcd8908664508bbed53cd56c2728daf4a7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-20 13:01:20.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-20 13:01:20.000000000","tz":-240},"subject":"Remove egg_info from setup.cfg","message":"Remove egg_info from setup.cfg\n\nThis was removed from cinder by change If7db350ea55a that removed\npy27 support, but we neglected to remove it from os-brick, and so it\ngot included in the python-3-only rbd-iscsi-client because its initial\nproject structure was based on os-brick.\n\nChange-Id: Ibbca458dd936f9ec8b93fb8d5d0bc9c1f05f5467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f5e6b43633b9993a8600660be9c42a3c84e20f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/f5e6b43633b9993a8600660be9c42a3c84e20f5a"}]},"branch":"refs/heads/master"},"1ecfac3da494fc4d593b324e9fb4fe1da52ec913":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-21 23:35:42.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/94/889094/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/94/889094/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/94/889094/2"}}},"commit":{"parents":[{"commit":"ec06206754c312af6d8beacad0dd3f804a16eee8","subject":"Set a language in sphinx doc config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/ec06206754c312af6d8beacad0dd3f804a16eee8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-20 13:01:20.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-11-21 23:35:42.000000000","tz":0},"subject":"Remove egg_info from setup.cfg","message":"Remove egg_info from setup.cfg\n\nThis was removed from cinder by change If7db350ea55a that removed\npy27 support, but we neglected to remove it from os-brick, and so it\ngot included in the python-3-only rbd-iscsi-client because its initial\nproject structure was based on os-brick.\n\nChange-Id: Ibbca458dd936f9ec8b93fb8d5d0bc9c1f05f5467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/1ecfac3da494fc4d593b324e9fb4fe1da52ec913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/1ecfac3da494fc4d593b324e9fb4fe1da52ec913"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
