)]}'
{"id":"openstack%2Frbd-iscsi-client~987459","triplet_id":"openstack%2Frbd-iscsi-client~master~Ifb891dbed0943c7c0532e0f6a9298cd200b59ade","project":"openstack/rbd-iscsi-client","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-11 20:23:31.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ifb891dbed0943c7c0532e0f6a9298cd200b59ade","subject":"Bump hacking","status":"NEW","created":"2026-05-06 08:25:41.000000000","updated":"2026-05-11 20:23:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42613857d95956cd367df8d71921e9e98283e319","_number":987459,"virtual_id_number":987459,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-06 08:50:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 08:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b01356d61ec006808e642401b499ef80e410602d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 08:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb6bf0d6dcd686614965834a01a5da8dbb4d5764","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 08:30:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e5145a93b024c7845de3f877fdbe2d7ee8484938","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 08:30:10.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"9a4406395e540e2839b04ee512c96e6e71cd228f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 08:38:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a795bc720c3947c387d69ab6d10e7436\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c268be56808348ae87074168796f7568 : SUCCESS in 2m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/db6b03e4c0c649f299c82afd09b682c8 : SUCCESS in 2m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b388da22392f4061a650be11a00be8bd : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9a4dbc7f82444c9ae3e7f019efa1c9 : FAILURE in 2m 40s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/09673f19ae1546e3ac60961d4790ff07 : SUCCESS in 49s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/7dfa3a4fc38a45449f793aef689b5185 : FAILURE in 1m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64c93786ef3d425fb8aeaedc9780e408 : FAILURE in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccbfe996c4694a24979d231fc2911952 : SUCCESS in 1m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/938306ac183f4dfd8abce68525ebe17c : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82d80c82acf640ae91e81fcdf1efc805 : FAILURE in 3m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file rbd_iscsi_client/tests/test_client_request.py","accounts_in_message":[],"_revision_number":2},{"id":"ca842eed7436da0a9c5fea6d9cf89f910a6c0491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-06 08:40:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a5d42d61c6426f6c6ec784e502f40b7c021ece96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 08:50:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0fc9698697f4793a293dfff243b3fa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe4eb999dd543e8adce2705ba0bb93e : SUCCESS in 4m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/37d5e89e899b4b49aabe9289de6a01b0 : SUCCESS in 1m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e8c207402da4b36ad424b4919386c88 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127f34f80d1e4e8fa1eb440801690b18 : SUCCESS in 2m 55s\n- rbd-iscsi-client-py36 https://zuul.opendev.org/t/openstack/build/e8d8c2f1bab942e5bd5fb301b57e988a : SUCCESS in 39s\n- rbd-iscsi-client-py38 https://zuul.opendev.org/t/openstack/build/223dfdb3f12c49fc903053acf7a53000 : FAILURE in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03346aa2080b46a0a5c2236e9fbc1b68 : FAILURE in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19e9c495bdf44fb3a064122c89f7940c : SUCCESS in 1m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7229ebd0e1724da8b619b57bfd6349d5 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d623010bc34347548a7518843c2c3237 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22883adac3fcccdc8430e6685ce4a0769c615867","revisions":{"8b2fe49763f86b9bee31fc9a3c9873ed4fc80189":{"kind":"REWORK","_number":1,"created":"2026-05-06 08:25:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/59/987459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/59/987459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/1"}}},"commit":{"parents":[{"commit":"a9e42f55790e45d9f90db2c84328862e43ba2bd7","subject":"Update CI for Dalmatian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/a9e42f55790e45d9f90db2c84328862e43ba2bd7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 08:25:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 08:25:41.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old, new version fix problems with flake8.\n\nChange-Id: Ifb891dbed0943c7c0532e0f6a9298cd200b59ade\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/8b2fe49763f86b9bee31fc9a3c9873ed4fc80189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/8b2fe49763f86b9bee31fc9a3c9873ed4fc80189"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"d6e2de4bccafa7d817fbcf8856b6f476361def3c":{"kind":"REWORK","_number":2,"created":"2026-05-06 08:30:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/59/987459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/59/987459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/2"}}},"commit":{"parents":[{"commit":"a9e42f55790e45d9f90db2c84328862e43ba2bd7","subject":"Update CI for Dalmatian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/a9e42f55790e45d9f90db2c84328862e43ba2bd7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 08:25:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 08:30:00.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old, new version fix problems with flake8.\n\nChange-Id: Ifb891dbed0943c7c0532e0f6a9298cd200b59ade\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/d6e2de4bccafa7d817fbcf8856b6f476361def3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/d6e2de4bccafa7d817fbcf8856b6f476361def3c"}]},"branch":"refs/heads/master"},"22883adac3fcccdc8430e6685ce4a0769c615867":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-06 08:40:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/59/987459/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rbd-iscsi-client","ref":"refs/changes/59/987459/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rbd-iscsi-client refs/changes/59/987459/3"}}},"commit":{"parents":[{"commit":"ec96fd42a53c3e47f6cd690607e88725584198d6","subject":"Replace assertRaisesRegexp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/ec96fd42a53c3e47f6cd690607e88725584198d6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 08:25:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-06 08:40:33.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old, new version fix problems with flake8.\n\nChange-Id: Ifb891dbed0943c7c0532e0f6a9298cd200b59ade\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/22883adac3fcccdc8430e6685ce4a0769c615867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rbd-iscsi-client/commit/22883adac3fcccdc8430e6685ce4a0769c615867"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d6e2de4bccafa7d817fbcf8856b6f476361def3c","theirs":"ec96fd42a53c3e47f6cd690607e88725584198d6","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
