)]}'
{"id":"openstack%2Frelease-test~720176","triplet_id":"openstack%2Frelease-test~master~Id3396d31585787fa6e43d9c2857fe1df1c2b6684","project":"openstack/release-test","branch":"master","topic":"mirror-test","hashtags":[],"change_id":"Id3396d31585787fa6e43d9c2857fe1df1c2b6684","subject":"Do not reference non-existent LP project","status":"MERGED","created":"2020-04-15 10:29:51.000000000","updated":"2020-04-15 10:30:50.000000000","submitted":"2020-04-15 10:30:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720176-1586946650576-79213a0a","meta_rev_id":"e2508dd3cf9824d9a604da3b4699cac435889c06","_number":720176,"virtual_id_number":720176,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 10:30:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-04-15 10:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-04-15 10:30:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved by PTL"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 10:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39ba2aa567cc3dca34c79793fe1e13d97eb0d78a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-15 10:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c313237ee6e4c8ffe10496656e88f200a4c8bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:30:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/b8b0cc5a86e04ec2aa695275380de452 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"b84ae21c28ff3d646c925f4212bd0e73b7ee2847","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-15 10:30:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSelf-approving release-test change to trigger GitHub mirroring job.","accounts_in_message":[],"_revision_number":1},{"id":"1b6cdeceb53257241cde15b68e0aa5b02245e980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:30:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95b0d2ea6865662eb8814cc002502d54443415d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:30:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ffacbf2a7edf6379edd6d03e9b31a1c0cbcce11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:30:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/8f9cce9f6bd84a9393930b0a8f8df04c : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee8eb7a344fd1df3cae46e9de57d77b635f2aa7d","revisions":{"ee8eb7a344fd1df3cae46e9de57d77b635f2aa7d":{"kind":"REWORK","_number":1,"created":"2020-04-15 10:29:51.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/76/720176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/release-test","ref":"refs/changes/76/720176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/release-test refs/changes/76/720176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/release-test refs/changes/76/720176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/release-test refs/changes/76/720176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/release-test refs/changes/76/720176/1"}}},"commit":{"parents":[{"commit":"d6c3d7e06156ceede7f6429d872d094f1ac6bf17","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/d6c3d7e06156ceede7f6429d872d094f1ac6bf17"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-15 10:28:24.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-15 10:29:49.000000000","tz":120},"subject":"Do not reference non-existent LP project","message":"Do not reference non-existent LP project\n\nFix CONTRIBUTING.rst so that it does not point to the non-existent\n\u0027release-test\u0027 project on Launchpad.\n\nChange-Id: Id3396d31585787fa6e43d9c2857fe1df1c2b6684\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/ee8eb7a344fd1df3cae46e9de57d77b635f2aa7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/ee8eb7a344fd1df3cae46e9de57d77b635f2aa7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
