)]}'
{"id":"openstack%2Frelease-test~720183","triplet_id":"openstack%2Frelease-test~master~Ic78fd44fd9e59cea202a558359e2987e27453b07","project":"openstack/release-test","branch":"master","topic":"mirror-test","hashtags":[],"change_id":"Ic78fd44fd9e59cea202a558359e2987e27453b07","subject":"Fix Its spelling","status":"MERGED","created":"2020-04-15 10:40:36.000000000","updated":"2020-04-15 10:55:25.000000000","submitted":"2020-04-15 10:55:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720183-1586948125961-79213a0a","meta_rev_id":"5f0e5b87018909eec1e57cd082454abc7ce7bda5","_number":720183,"virtual_id_number":720183,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 10:55:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-04-15 10:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-04-15 10:41:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved by PTL"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 10:55:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84e074e6f414efd60124296bb0387130a0f112bc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-15 10:40:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4aec0690d078a35e69e784650592cd1e34148dbc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-15 10:41:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSelf-approving to trigger GitHub replication test","accounts_in_message":[],"_revision_number":1},{"id":"a47917cec63b50402a2aeaa00d98fe34421a58f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:49:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/be337c3ea2f9404aa00a78ac3b650bc0 : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5cee7a1e206d44838e52d5dc091e35cd : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e9016660e17ed5ce403f761203c7f5984d14b35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:49:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f2a292a0db69420cf7f744db59d047121fffbec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:55:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d0d0565d22613f6242d0e2c1b1305d0c603137d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:55:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/93103b9137b14d56a755a7df1daaac4d : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5f63efc7d45420da89293a76db1acfc : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"600c783cc95005ba70a6aef1847b21587849de33","revisions":{"600c783cc95005ba70a6aef1847b21587849de33":{"kind":"REWORK","_number":1,"created":"2020-04-15 10:40:36.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/83/720183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/release-test","ref":"refs/changes/83/720183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/release-test refs/changes/83/720183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/release-test refs/changes/83/720183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/release-test refs/changes/83/720183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/release-test refs/changes/83/720183/1"}}},"commit":{"parents":[{"commit":"ee8eb7a344fd1df3cae46e9de57d77b635f2aa7d","subject":"Do not reference non-existent LP project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/ee8eb7a344fd1df3cae46e9de57d77b635f2aa7d"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-15 10:39:56.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-15 10:39:56.000000000","tz":120},"subject":"Fix Its spelling","message":"Fix Its spelling\n\nAnother inconsequential change to test GitHub mirroring.\n\nChange-Id: Ic78fd44fd9e59cea202a558359e2987e27453b07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/600c783cc95005ba70a6aef1847b21587849de33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/600c783cc95005ba70a6aef1847b21587849de33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
