)]}'
{"id":"openstack%2Frelease-test~984047","triplet_id":"openstack%2Frelease-test~master~Ie6ea041cd5424a01984a6fc013287eaef13e5374","project":"openstack/release-test","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-04-10 17:07:22.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-10 16:59:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ie6ea041cd5424a01984a6fc013287eaef13e5374","subject":"Introduce pyproject.toml","status":"MERGED","created":"2026-04-10 15:01:52.000000000","updated":"2026-04-10 17:19:52.000000000","submitted":"2026-04-10 17:19:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984047","meta_rev_id":"1d83803a3e10704254c4bf7dddb05c2ba95b559a","_number":984047,"virtual_id_number":984047,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-10 17:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-04-10 17:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-04-10 16:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-04-10 17:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved by PTL"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 15:11:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-10 15:15:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"49548afcce0d49f67031f3d7e240ba03c05eac8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 15:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"063f3bf9a4952cd2c14b26b0ec41e17e90a8adff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:11:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638d1c8fcd524448900b650848692db2\n\n- noop https://zuul.opendev.org/t/openstack/build/faf6cf2457394e9a97100c0786c3c1b7 : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36121870e347480c9528f041e71aea79 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"40020eb2f0d41184a82f00af23b1324896240d4a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 15:11:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db21acc68cc8bf51bf0a713831f13ce688e40d24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85ef2971b15776e00c6027944fb43b1c63d5687c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 15:15:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc6f9f5dc9f4f5b13fce5b4a145c542ad180da9a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 15:16:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1dca011297a50857c17195ed3d09c1469641fbfb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 15:17:34.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"9e3e6c80f08a6afb15ee06144c7fb21840d089f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:17:35.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"695414f11d59d107725e3d08e708a004937321c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 15:42:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":2},{"id":"e0d33d25eaa60727bdea8635539ae1baa33a08bc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 15:44:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63bf4cee4e38dd5d514e16eb0d30e068fb3bd779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:51:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5157e71b8d6b4e9b93ffc7c1a3c38711\n\n- noop https://zuul.opendev.org/t/openstack/build/f9acd3b4fe4a45da9610d53810cc050e : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6163432a0de24ec6ba7ee72c718f4e6e : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7af0ccb86cd2694b8348a1517822f22bb62b1e79","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 16:26:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c19d13e800671175b34272103fc1cf5384342433","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 16:58:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a96d026d803a97b0e0598c720ed3194ed8466603","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 16:59:04.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"277674ea78721beb1534e4047bd1b7dec9572180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 17:00:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a482d90289d4a4eb1cf1f67add90317\n\n- noop https://zuul.opendev.org/t/openstack/build/d7bceb22a6634f68bdaa565d63befe4f : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/301b69dfc84d459ea42aeffdedacca91 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":3},{"id":"656cc82b89440dd95d85012f7827114229d6cd09","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-10 17:07:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68d84757c3a8681e9a8c97a9f7f18cc4a8ed60fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 17:08:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1eb0ace56561bf5da048cd7ffcc836eae00bcb78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 17:19:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0413112c7ab84ed1a9470daa24d66b43\n\n- noop https://zuul.opendev.org/t/openstack/build/a95dbc6eb0f2498799821ab623998014 : SUCCESS in 0s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1cdd0b260c4463db6eaa37ec0fdece9 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":3},{"id":"1d83803a3e10704254c4bf7dddb05c2ba95b559a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 17:19:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a802a845378a1c63037b5c4ca8632e34e2ff0053","revisions":{"e6e5d9441c7ee707ccc2b8931d178751de2be5ef":{"kind":"REWORK","_number":1,"created":"2026-04-10 15:01:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/984047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/release-test","ref":"refs/changes/47/984047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/release-test refs/changes/47/984047/1"}}},"commit":{"parents":[{"commit":"9dfaf3fcb8033eafca03b7de7fa65a656c61a695","subject":"Bump required python3 version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/9dfaf3fcb8033eafca03b7de7fa65a656c61a695"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-10 14:59:26.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-10 15:01:49.000000000","tz":120},"subject":"Introduce pyproject.toml","message":"Introduce pyproject.toml\n\nChange-Id: Ie6ea041cd5424a01984a6fc013287eaef13e5374\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/e6e5d9441c7ee707ccc2b8931d178751de2be5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/e6e5d9441c7ee707ccc2b8931d178751de2be5ef"}]},"branch":"refs/heads/master"},"43dc89517fd111d770c5294a0dc3f6ee712e482f":{"kind":"REWORK","_number":2,"created":"2026-04-10 15:42:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/984047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/release-test","ref":"refs/changes/47/984047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/release-test refs/changes/47/984047/2"}}},"commit":{"parents":[{"commit":"9dfaf3fcb8033eafca03b7de7fa65a656c61a695","subject":"Bump required python3 version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/9dfaf3fcb8033eafca03b7de7fa65a656c61a695"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-10 14:59:26.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-10 15:24:29.000000000","tz":120},"subject":"Introduce pyproject.toml","message":"Introduce pyproject.toml\n\nThis patch adds pyproject.toml and updated setup.cfg to have newer\npython versions listed as classifiers.\n\nAssisted-by: Claude Sonnet 4.6\nChange-Id: Ie6ea041cd5424a01984a6fc013287eaef13e5374\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/43dc89517fd111d770c5294a0dc3f6ee712e482f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/43dc89517fd111d770c5294a0dc3f6ee712e482f"}]},"branch":"refs/heads/master"},"a802a845378a1c63037b5c4ca8632e34e2ff0053":{"kind":"REWORK","_number":3,"created":"2026-04-10 16:58:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/984047/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/release-test","ref":"refs/changes/47/984047/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/release-test refs/changes/47/984047/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/release-test refs/changes/47/984047/3"}}},"commit":{"parents":[{"commit":"9dfaf3fcb8033eafca03b7de7fa65a656c61a695","subject":"Bump required python3 version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/9dfaf3fcb8033eafca03b7de7fa65a656c61a695"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-10 14:59:26.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-10 16:57:27.000000000","tz":120},"subject":"Introduce pyproject.toml","message":"Introduce pyproject.toml\n\nThis patch adds pyproject.toml and removes the unnecessary parts of\nsetup.cfg now that things are moved to pyproject.toml.\n\nAssisted-by: Claude Sonnet 4.6\nChange-Id: Ie6ea041cd5424a01984a6fc013287eaef13e5374\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/a802a845378a1c63037b5c4ca8632e34e2ff0053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/release-test/commit/a802a845378a1c63037b5c4ca8632e34e2ff0053"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"Marked as approved by the PTL","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
