)]}'
{"id":"openstack%2Freleases~1001303","triplet_id":"openstack%2Freleases~master~I2509c3b2dc42e6df8c3021639e1fc7a44edfab56","project":"openstack/releases","branch":"master","topic":"hibiscus-final-non-client-libs","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2026-08-25 07:39:41.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2026-08-25 07:33:08.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-25 07:39:41.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-08-25 07:39:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2509c3b2dc42e6df8c3021639e1fc7a44edfab56","subject":"Release final mistral-extra for 2026.2 Hibiscus","status":"MERGED","created":"2026-08-18 10:39:18.000000000","updated":"2026-08-25 07:39:41.000000000","submitted":"2026-08-25 07:39:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1001303-hibiscus-final-non-client-libs","meta_rev_id":"66f16e3d23b5e79a5c373198ba2f09c14c60998c","_number":1001303,"virtual_id_number":1001303,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-25 07:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2026-08-25 07:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2026-08-25 07:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2026-08-25 07:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-18 10:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-18 10:53:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-21 14:21:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-08-21 14:21:18.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2026-08-24 15:36:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"9d6a4d774b022b7c7b084ff209494fc8a85a8e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-18 10:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6fb0148a2d562d807040d1f2c85792d998fcdde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 10:53:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ef812111074e868cd9764dc890a6a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec92e84c3c1a497d91202277007a6c8a : SUCCESS in 6m 55s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/151eb74070764b4088f040aad3d20f87 : SUCCESS in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2ca425d132654ef49db451c9c17eb85a : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"292c280ba01b8ef6fd17e2036ba843b050f8ffd7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-21 14:30:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf2555c6f3e39a343fb17fb257d86ae9f818307a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-24 15:35:59.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c9a4f715952a648fc8173a776561ec970d82268c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-24 15:36:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46747116de4c509642f11b221683a6b5e6064579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:12:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b347434e75242d2a0fec7ed4696aeb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4642ad035f46c4811d5e6388a6cf3e : SUCCESS in 7m 24s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/527558b1c7e54173af3debc2f9793088 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/25450e6811ed43709800590e178fe28d : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"62cdbcd8e99e242e96ce3a579125c0de0873baad","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-25 07:33:08.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd4af1ed99a695799e1f8d1bd4cf9f4b0379fdbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:33:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f58fe7c181affc0b2f1416de6182ad96b3c83da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:39:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dcac801f66445929f99ff226311bf4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5313808f1de44a92b27d5e17bc5ab2b3 : SUCCESS in 6m 01s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/99112e49b35e4f628ba9e5cda5234f18 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"66f16e3d23b5e79a5c373198ba2f09c14c60998c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:39:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b1488cd6590ba8980c4541686755d7d05aaff3c","revisions":{"22f0e7cb7d1d86fc465ae4ace6290ba9a2d0b848":{"kind":"REWORK","_number":1,"created":"2026-08-18 10:39:18.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/03/1001303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/1001303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/1001303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/1001303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/1001303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/1001303/1"}}},"commit":{"parents":[{"commit":"f2f9020287b12a0b0f3ac054a08541bfd514ca4e","subject":"Merge \"Release pbr 7.1.0 version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2f9020287b12a0b0f3ac054a08541bfd514ca4e"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:39:14.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:39:14.000000000","tz":120},"subject":"Release final mistral-extra for 2026.2 Hibiscus","message":"Release final mistral-extra for 2026.2 Hibiscus\n\nThis proposes a release for mistral-extra to get any unreleased\nchanges out prior to library freeze for 2026.2 Hibiscus going into\neffect.\n\nIf the team is ready to proceed, then please leave a +1 to acknowledge\nthis, so that we can process this right away.\n\nIf the team is working on any final merges, leave a -1 with some detail\non the status and update the patch with the new commit hash to use when\nready.\n\nEither way, please do respond on this patch before EOD Thursday, August\n20th. If the patch gets no response by then, then we\u0027ll consider this\nis OK to be released.\n\n$ git log --oneline --no-merges 16.0.1..2490823\n792adda Fix uncaught direct string interpolation in log\n4eab3d6 Remove vitrage actions\n3dbca9e Drop implicit dependency on importlib_metadata\n20d4b4e tox: Drop redundant override of install_command\ne938443 Restore explicit package list for module detection\n7fbe54a Remove heat services from example local.conf\n47e59c1 Update packaging configuration\n02b9d2e Drop Python 3.10\nfea34cf Remove delete steps for pyc files\nf4d291c Update master for stable/2026.1\n86fee70 Remove unused sphinx extensions\n\nChange-Id: I2509c3b2dc42e6df8c3021639e1fc7a44edfab56\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22f0e7cb7d1d86fc465ae4ace6290ba9a2d0b848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22f0e7cb7d1d86fc465ae4ace6290ba9a2d0b848"}]},"branch":"refs/heads/master"},"7b1488cd6590ba8980c4541686755d7d05aaff3c":{"kind":"REWORK","_number":2,"created":"2026-08-24 15:35:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/03/1001303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/1001303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/1001303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/1001303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/1001303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/1001303/2"}}},"commit":{"parents":[{"commit":"f2f9020287b12a0b0f3ac054a08541bfd514ca4e","subject":"Merge \"Release pbr 7.1.0 version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f2f9020287b12a0b0f3ac054a08541bfd514ca4e"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:39:14.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-24 15:35:59.000000000","tz":0},"subject":"Release final mistral-extra for 2026.2 Hibiscus","message":"Release final mistral-extra for 2026.2 Hibiscus\n\nThis proposes a release for mistral-extra to get any unreleased\nchanges out prior to library freeze for 2026.2 Hibiscus going into\neffect.\n\nIf the team is ready to proceed, then please leave a +1 to acknowledge\nthis, so that we can process this right away.\n\nIf the team is working on any final merges, leave a -1 with some detail\non the status and update the patch with the new commit hash to use when\nready.\n\nEither way, please do respond on this patch before EOD Thursday, August\n20th. If the patch gets no response by then, then we\u0027ll consider this\nis OK to be released.\n\n$ git log --oneline --no-merges 16.0.1..2490823\n792adda Fix uncaught direct string interpolation in log\n4eab3d6 Remove vitrage actions\n3dbca9e Drop implicit dependency on importlib_metadata\n20d4b4e tox: Drop redundant override of install_command\ne938443 Restore explicit package list for module detection\n7fbe54a Remove heat services from example local.conf\n47e59c1 Update packaging configuration\n02b9d2e Drop Python 3.10\nfea34cf Remove delete steps for pyc files\nf4d291c Update master for stable/2026.1\n86fee70 Remove unused sphinx extensions\n\nChange-Id: I2509c3b2dc42e6df8c3021639e1fc7a44edfab56\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b1488cd6590ba8980c4541686755d7d05aaff3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b1488cd6590ba8980c4541686755d7d05aaff3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
