)]}'
{"id":"openstack%2Freleases~1001322","triplet_id":"openstack%2Freleases~master~I0727b731c4b0e89a6c8501ea650a713d060f4d70","project":"openstack/releases","branch":"master","topic":"hibiscus-final-non-client-libs","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-08-25 07:41:58.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2026-08-25 07:33:56.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-25 07:41:58.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-25 07:41:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0727b731c4b0e89a6c8501ea650a713d060f4d70","subject":"Release final glance_store for 2026.2 Hibiscus","status":"MERGED","created":"2026-08-18 10:55:38.000000000","updated":"2026-08-25 07:41:58.000000000","submitted":"2026-08-25 07:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1001322-hibiscus-final-non-client-libs","meta_rev_id":"9755d256e4e335d45e6dc7f2fe5bf33b8c859ebc","_number":1001322,"virtual_id_number":1001322,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-25 07:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2026-08-25 07:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2026-08-25 07:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2026-08-25 07:33:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-18 10:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-18 11:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-21 14:19:38.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-08-21 14:19:39.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-08-24 15:10:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"6ff8403916b27a789ecb118a2e61f74a29ce8a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-18 10:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd249394617844448a07d53efa1fa00cea35aba8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-18 10:56:24.000000000","message":"Topic set to hibiscus-final-non-client-libs","accounts_in_message":[],"_revision_number":1},{"id":"88f235b09fdeade29bf2a212961cf20fa1cd6276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 11:11:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b130c4847808431ba6902807de4d337f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49c914aa6eb4beba35fa249ca47e4dc : SUCCESS in 7m 03s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1599bd7d64c04e53a1d0714f6dc9e68f : SUCCESS in 6m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8e05a8b57f9840f3894398171fbf14d5 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1574f0dca0dd442a32d149180fe5bf5a2e73bfb7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-21 14:29:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a5201a458c8fa576d9756973576a074eb132fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-24 14:05:01.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"267710d6d1a5e7d87cf099a0bba40f656636a876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 14:14:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16f005fb4ee447178e3e7b4c512d5172\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52577bbb4d1449a0bac22374e24ea6a8 : SUCCESS in 4m 11s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/df01ecb78b1e4ae8aca49128b00a6460 : FAILURE in 4m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/89e1e55670ee4b10b835685689913315 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1152196d6743057f2045e6595e76ddaf4bd63658","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-24 15:10:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8edafa4da9f6736db2f15e91fcbbeb4f8d032f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-24 15:11:08.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"07ad8019fd56107bb7045946f20b3c913db5ee43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:38:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76aa4e89a2034f768a4733163d3d158a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98fd2789ff264d128e5e094d6509154c : SUCCESS in 6m 58s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e668ccd8b948442991dbdffffdcdefbe : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85a5d962a8ed400fa326a1d8d2699dd1 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"6afca23440233253933151444636bff30f8c7e1f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-08-25 07:33:56.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73d68c15e7eec1ceeab302531ce4179d3d5e1a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:34:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e432c09a0c6ac14fc5e13dcd88e753be2202a12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:41:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e2f1fdbd8342f28446e096c006cbfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805d72c94f6a479a832852846121b60f : SUCCESS in 6m 23s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/df586183cd8e4d719f2aa6891c0a6e26 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9755d256e4e335d45e6dc7f2fe5bf33b8c859ebc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:41:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b4b9cfc1002c1bbfe557873492e605a99cef6a78","revisions":{"38856869ce4d10cb1208ab0ece4b0f8cd915433a":{"kind":"REWORK","_number":1,"created":"2026-08-18 10:55:38.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/22/1001322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/1001322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/1001322/1"}}},"commit":{"parents":[{"commit":"f5cbd773fd453f59d7002a0c34c5871d71ed8868","subject":"Release Rally 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5cbd773fd453f59d7002a0c34c5871d71ed8868"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:53:24.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:53:24.000000000","tz":120},"subject":"Release final glance_store for 2026.2 Hibiscus","message":"Release final glance_store for 2026.2 Hibiscus\n\nThis proposes a release for glance_store to get any unreleased\nchanges out prior to library freeze for 2026.2 Hibiscus going into\neffect.\n\nIf the team is ready to proceed, then please leave a +1 to acknowledge\nthis, so that we can process this right away.\n\nIf the team is working on any final merges, leave a -1 with some detail\non the status and update the patch with the new commit hash to use when\nready.\n\nEither way, please do respond on this patch before EOD Thursday, August\n20th. If the patch gets no response by then, then we\u0027ll consider this\nis OK to be released.\n\nChanges between 5.6.0 and 7756488:\n* 214908d 2026-08-12 14:02:52 +0000 reno: Update master for unmaintained/2023.1\n* 7739485 2026-08-09 03:27:49 +0000 Imported Translations from Zanata\n\nChange-Id: I0727b731c4b0e89a6c8501ea650a713d060f4d70\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38856869ce4d10cb1208ab0ece4b0f8cd915433a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38856869ce4d10cb1208ab0ece4b0f8cd915433a"}]},"branch":"refs/heads/master"},"f44770883dfcd2d0c233c49e5715cd3bc8856655":{"kind":"REWORK","_number":2,"created":"2026-08-24 14:05:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/22/1001322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/1001322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/1001322/2"}}},"commit":{"parents":[{"commit":"f5cbd773fd453f59d7002a0c34c5871d71ed8868","subject":"Release Rally 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5cbd773fd453f59d7002a0c34c5871d71ed8868"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:53:24.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-24 14:05:01.000000000","tz":0},"subject":"Release final glance_store for 2026.2 Hibiscus","message":"Release final glance_store for 2026.2 Hibiscus\n\nThis proposes a release for glance_store to get any unreleased\nchanges out prior to library freeze for 2026.2 Hibiscus going into\neffect.\n\nIf the team is ready to proceed, then please leave a +1 to acknowledge\nthis, so that we can process this right away.\n\nIf the team is working on any final merges, leave a -1 with some detail\non the status and update the patch with the new commit hash to use when\nready.\n\nEither way, please do respond on this patch before EOD Thursday, August\n20th. If the patch gets no response by then, then we\u0027ll consider this\nis OK to be released.\n\nChanges between 5.6.0 and 7756488:\n* 214908d 2026-08-12 14:02:52 +0000 reno: Update master for unmaintained/2023.1\n* 7739485 2026-08-09 03:27:49 +0000 Imported Translations from Zanata\n\nChange-Id: I0727b731c4b0e89a6c8501ea650a713d060f4d70\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f44770883dfcd2d0c233c49e5715cd3bc8856655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f44770883dfcd2d0c233c49e5715cd3bc8856655"}]},"branch":"refs/heads/master"},"b4b9cfc1002c1bbfe557873492e605a99cef6a78":{"kind":"REWORK","_number":3,"created":"2026-08-24 15:11:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/22/1001322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/1001322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/1001322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/1001322/3"}}},"commit":{"parents":[{"commit":"f5cbd773fd453f59d7002a0c34c5871d71ed8868","subject":"Release Rally 5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f5cbd773fd453f59d7002a0c34c5871d71ed8868"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2026-08-18 10:53:24.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-24 15:11:08.000000000","tz":0},"subject":"Release final glance_store for 2026.2 Hibiscus","message":"Release final glance_store for 2026.2 Hibiscus\n\nThis proposes a release for glance_store to get any unreleased\nchanges out prior to library freeze for 2026.2 Hibiscus going into\neffect.\n\nIf the team is ready to proceed, then please leave a +1 to acknowledge\nthis, so that we can process this right away.\n\nIf the team is working on any final merges, leave a -1 with some detail\non the status and update the patch with the new commit hash to use when\nready.\n\nEither way, please do respond on this patch before EOD Thursday, August\n20th. If the patch gets no response by then, then we\u0027ll consider this\nis OK to be released.\n\nChanges between 5.6.0 and 7756488:\n* 214908d 2026-08-12 14:02:52 +0000 reno: Update master for unmaintained/2023.1\n* 7739485 2026-08-09 03:27:49 +0000 Imported Translations from Zanata\n\nChange-Id: I0727b731c4b0e89a6c8501ea650a713d060f4d70\nSigned-off-by: Thierry Carrez \u003cthierry@openstack.org\u003e\nGenerated-By: openstack/releases:tools/process_auto_releases.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4b9cfc1002c1bbfe557873492e605a99cef6a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4b9cfc1002c1bbfe557873492e605a99cef6a78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
