)]}'
{"id":"openstack%2Freleases~1003754","triplet_id":"openstack%2Freleases~master~Iebe065a30946cda548ec22f38e00f959d6fb3059","project":"openstack/releases","branch":"master","topic":"tosca-parser-transition-independent","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-09-08 13:19:54.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2026-09-08 13:19:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebe065a30946cda548ec22f38e00f959d6fb3059","subject":"[tacker] Force-release tosca-parser to hibiscus","status":"MERGED","created":"2026-09-03 11:41:50.000000000","updated":"2026-09-08 13:19:54.000000000","submitted":"2026-09-08 13:19:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003754-tosca-parser-transition-independent","meta_rev_id":"d0a57c3512e2470cec80ea00a0da840d56a9c0b3","_number":1003754,"virtual_id_number":1003754,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 13:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-08 13:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-08 13:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-09-08 06:15:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 11:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 12:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 06:58:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2026-09-08 13:04:40.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"376193ede8e5a4ce4a5c5fb61797d5d4c3a1577b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-03 11:41:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"775628fecdcd6bf9e9a5ad8e284412412eca9669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:11:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff2146328e1248c8a0e827067f04d2bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af36aa5c96264a32993860073cb49191 : SUCCESS in 9m 56s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/27e740ae8f414753a7dd96c943af66a7 : FAILURE in 8m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/12f719ff38ec41e6b8f986e722e09c8a : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e7803f723a4f385ccf6d42bc33e4763ce0a44ab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-03 12:36:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e31a91e9feaf72a907c5ebfff81cddc6e8bd4d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:47:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd7c83dc4b3b4f02aaacf582ff03612f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c7be99177b41d9b8dd5ed7854336c4 : SUCCESS in 7m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/0f22017d88904cc0844e72e057daab9a : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1d4147ee4a384969b16dea4f308b830f : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"388bfdfb4d71c8289b2dcea6a63616bed0619e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-08 06:12:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d319d23932a938e23acb45b33acc89776120e90c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-08 06:15:10.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"7e1b021aec256c810182505316e99617d6e1cef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 06:19:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e37956841b549a0b7ee60f78a5afdc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad964210916f4e34b2e317a5ee955f92 : SUCCESS in 4m 34s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/a1f9f9e761dd40d4b2462eb1df8c8851 : SUCCESS in 2m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9d7e90e233824a3d9831c911a914f676 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f6759e89a211a2018b29e963fa4ad8541b39309e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-08 13:04:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"99ec616cd71073dfddf696dd074bc32145de08c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 13:04:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ee860f750526aabb2f02729727efba70a8bf2f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 13:19:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1c49b5123a4ba08f30d474ccc7f1ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44b48f3f70147c3a4c44562917ebd66 : SUCCESS in 6m 50s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/825a57b538b642d1b8050c34d59e3701 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d0a57c3512e2470cec80ea00a0da840d56a9c0b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 13:19:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf14e286aff0da03f1e84e6960207191ede29479","revisions":{"66c1fa6ec70a4931c9b84f35c5773d557e2b24ce":{"kind":"REWORK","_number":1,"created":"2026-09-03 11:41:50.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/1003754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/1003754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/1003754/1"}}},"commit":{"parents":[{"commit":"b483a5324b6a3d2ee5eae2376c82135cce699c2c","subject":"Merge \"process: Clarify requirements team communication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b483a5324b6a3d2ee5eae2376c82135cce699c2c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-03 11:30:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-03 11:39:42.000000000","tz":120},"subject":"[tacker] Move tosca-parser to cycle independent model","message":"[tacker] Move tosca-parser to cycle independent model\n\ntosca-parser was not released in the current series, either it is\nmature / stable enough and we don\u0027t need any change in it, or it\nshould be retired if no one need it. This patch moves it to cycle\nindependent release model, which means it can be released any time\nwhen needed, but no need to force release during development cycles.\n\nNote, that by changing release model\n* it won\u0027t be possible to cut \"stable\" releases as there won\u0027t be new\n  stable/202*.* branches cut per development cylce\n* to avoid issues with missing stable releases: the code base *must*\n  comply with all open branches where the code is used (i.e. if\n  tosca-parser is used in 2026.2 branch, then the code on tip of the\n  master branch should be compatible with supported python version of\n  2026.2 release; etc.)\n\nChange-Id: Iebe065a30946cda548ec22f38e00f959d6fb3059\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66c1fa6ec70a4931c9b84f35c5773d557e2b24ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66c1fa6ec70a4931c9b84f35c5773d557e2b24ce"}]},"branch":"refs/heads/master"},"5e7b508c31ab3a7c597f33dbd01a9c4cacce1659":{"kind":"REWORK","_number":2,"created":"2026-09-03 12:36:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/1003754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/1003754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/1003754/2"}}},"commit":{"parents":[{"commit":"b483a5324b6a3d2ee5eae2376c82135cce699c2c","subject":"Merge \"process: Clarify requirements team communication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b483a5324b6a3d2ee5eae2376c82135cce699c2c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-03 11:30:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-03 12:36:29.000000000","tz":120},"subject":"[tacker] Move tosca-parser to cycle independent model","message":"[tacker] Move tosca-parser to cycle independent model\n\ntosca-parser was not released in the current series, either it is\nmature / stable enough and we don\u0027t need any change in it, or it\nshould be retired if no one need it. This patch moves it to cycle\nindependent release model, which means it can be released any time\nwhen needed, but no need to force release during development cycles.\n\nNote, that by changing release model\n* it won\u0027t be possible to cut \"stable\" releases as there won\u0027t be new\n  stable/202*.* branches cut per development cylce\n* to avoid issues with missing stable releases: the code base *must*\n  comply with all open branches where the code is used (i.e. if\n  tosca-parser is used in 2026.2 branch, then the code on tip of the\n  master branch should be compatible with supported python version of\n  2026.2 release; etc.)\n\nChange-Id: Iebe065a30946cda548ec22f38e00f959d6fb3059\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e7b508c31ab3a7c597f33dbd01a9c4cacce1659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e7b508c31ab3a7c597f33dbd01a9c4cacce1659"}]},"branch":"refs/heads/master"},"bf14e286aff0da03f1e84e6960207191ede29479":{"kind":"REWORK","_number":3,"created":"2026-09-08 06:12:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/54/1003754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/1003754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/1003754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/1003754/3"}}},"commit":{"parents":[{"commit":"0eb23e0bc65d330231265c2e1b8acad42b40fa1b","subject":"Merge \"[tool] Fix branch id fetching in scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0eb23e0bc65d330231265c2e1b8acad42b40fa1b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-03 11:30:38.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-09-08 06:07:38.000000000","tz":120},"subject":"[tacker] Force-release tosca-parser to hibiscus","message":"[tacker] Force-release tosca-parser to hibiscus\n\ntosca-parser was not released in the current series and seems inactive.\nSince we are late in the project to either move to cycle independent\nrelease model, or retire it, as agreed during last week\u0027s release team\nmeeting was decided to force-release tosca-parser.\n\nChange-Id: Iebe065a30946cda548ec22f38e00f959d6fb3059\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf14e286aff0da03f1e84e6960207191ede29479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf14e286aff0da03f1e84e6960207191ede29479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
