)]}'
{"id":"openstack%2Freleases~1004929","triplet_id":"openstack%2Freleases~master~I63d6ecb7a13faaba2faef039776365424ea6f826","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-11 12:58:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63d6ecb7a13faaba2faef039776365424ea6f826","subject":"[Neutron] Release a new 2026.1 version 28.0.2","status":"MERGED","created":"2026-09-09 20:36:24.000000000","updated":"2026-09-11 12:58:28.000000000","submitted":"2026-09-11 12:58:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004929","meta_rev_id":"8c1731ecea0173b4f761f4d2a8da121ad4e36972","_number":1004929,"virtual_id_number":1004929,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 12:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-09-11 11:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-11 12:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-11 12:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-09-10 21:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"}],"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 20:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 09:01:26.000000000","updated_by":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"reviewer":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"state":"CC"},{"updated":"2026-09-11 11:03:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-11 12:40:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"02e60662b17ddd06c27cc78f6c05d96dff241206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-09 20:36:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"354eba27a28ca3d189a54de526fae07e80fe6fb4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-09 20:36:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d3430d85fe626589352e854d2a8f65beb2b4a4d2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-09 20:37:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa46f75ea0f15c229c41643d59297b19d84e2810","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 20:37:50.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ef99a10d7590355e5c4b5bbe48e4af6c69e0c67f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 20:58:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b5c2a78c2d44aeb6cc44c862a0ed79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950f2eafed684d248c47c3ec2e7b4755 : SUCCESS in 8m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/885638ff500b459c8b22676959a8ba0e : SUCCESS in 8m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/56a181b29b404dfd95807d4bb502a8b4 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0465524644e2f5d7f7e0173f60b3a3a6d80ad067","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-10 21:18:32.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f4a3f8b839c7f1b8fa01479e481073075c9ed123","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-10 21:18:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c6bf9a1f98a23ca1f6dbe7fc4ad516b2cf48fa90","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 21:19:27.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"873c6968afc382b5212f0c5dbc4a6848e6f38946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 21:31:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66c997f158ff4555b80affc62497360e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8984dc1f7224d28955fef4ba4d7ceb6 : SUCCESS in 7m 16s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6d04c7e7b3d2415da3c5845ebb799387 : SUCCESS in 8m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2184b09a80a3413a9d19c96342625795 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"419cc78a566740ea8fa71bc2b9fbecf584f52351","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-11 11:03:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e8d332997b94f137ce0ad3f9e83acccecf159914","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2026-09-11 12:40:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"24f55dc96763a3982faddd955f4c0031bbb39b5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 12:41:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78fc1dfe9458dec0352eaaa8f9ec4a62c341cac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 12:58:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab85189deb2f4b39b4ee9b4fe99c242a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea12cd64a6ff487382ea427c234ae940 : SUCCESS in 6m 32s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/5838b2b138944d198b5c4869db1b473e : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8c1731ecea0173b4f761f4d2a8da121ad4e36972","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 12:58:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e1eb79e00770b9c24dec3e5dbc696f0ff24fc507","revisions":{"19ae2b0ec3f35918f4db254c6c771f7762530b79":{"kind":"REWORK","_number":1,"created":"2026-09-09 20:36:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/29/1004929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/1004929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/1004929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/1004929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/1004929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/1004929/1"}}},"commit":{"parents":[{"commit":"a34b0e7784d9e35f06edc503ea494f762e2edc53","subject":"Merge \"Release heat-dashboard RC1 for 2026.2 Hibiscus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a34b0e7784d9e35f06edc503ea494f762e2edc53"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-09-09 17:16:15.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-09-09 17:16:15.000000000","tz":-240},"subject":"[Neutron] Release a new 2026.1 version 28.0.2","message":"[Neutron] Release a new 2026.1 version 28.0.2\n\n$ git log --oneline --no-merges 28.0.1..d4b67e1ca2\n74e7df6233 FIP \u0026 QoS: Check if QoS Policy is visible for user\nf502deee9c bgp: Do not learn routes from main BGP LRP\n05db9a2d55 DHCP: Clean HAProxy processes from ns during port_delete\ne161372955 Use RBAC API policy to create default security group\nf8c429d589 ml2: notify mechanism drivers in reverse order during port delete\n3f1ce7f950 [CI][FT] Tempoarary skip failing tests centos 9 stream\n794cfbe187 [iptables_hybrid] Fix conntrack zone not being restored on restart\n0e7eed7fcf Allow shrinking a segment range to an allocated boundary\nda18231f98 Do not update unmodified ACLs in logapi service\nc7d0dc15bf Fix accidental removal of router external gateway info\n70d1c4e72f Validate RBAC ``target_project`` as UUID or wildcard\n94b8d4b1d4 Switch from setting LSP.tag to LSP.tag_request\n186c1a18ca Prevent cross-project subnet onboard on shared networks\n6e325eff58 Reject overlapping network segment ranges for tunnel types\nfcd54c1183 quota: Fix reservation under-counting in ``get_total_reservations_map()``\n0b01b68f1d qos: Validate parent policy ownership in ``get_policy_rule``\n0c1b467798 ovn: Ignore network HCGs in ``HAChassisGroupRouterEvent``\nfe5a333412 Fix load_db_migration_drivers\n5ebbea7055 ovn: Skip LSP host info update for router ports\n6479d51d6b ovn: read LSP ``up`` state from IDL cache in ``update_lsp_host_info``\nbba77b6ece rpc/ovn: Use base OVN IDL for RPC workers, disable notification events\n2bc474a7db [functional fips] Switch venv prep to python3.12\n408d683d71 segments: Catch RP not found in ``_delete_nova_inventory``\n3f0e8a3327 ovn: Fix HA_Chassis_Group race condition in port create/update\nb234755092 [UT] Update base_mac config override\ncae08d7db8 bgp: Limit lrp_mac to the chassis\n\nChange-Id: I63d6ecb7a13faaba2faef039776365424ea6f826\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19ae2b0ec3f35918f4db254c6c771f7762530b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19ae2b0ec3f35918f4db254c6c771f7762530b79"}]},"branch":"refs/heads/master"},"e1eb79e00770b9c24dec3e5dbc696f0ff24fc507":{"kind":"REWORK","_number":2,"created":"2026-09-10 21:18:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/29/1004929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/1004929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/1004929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/1004929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/1004929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/1004929/2"}}},"commit":{"parents":[{"commit":"a34b0e7784d9e35f06edc503ea494f762e2edc53","subject":"Merge \"Release heat-dashboard RC1 for 2026.2 Hibiscus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a34b0e7784d9e35f06edc503ea494f762e2edc53"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-09-09 17:16:15.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-09-10 21:18:30.000000000","tz":0},"subject":"[Neutron] Release a new 2026.1 version 28.0.2","message":"[Neutron] Release a new 2026.1 version 28.0.2\n\n$ git log --oneline --no-merges 28.0.1..d87ec8ef84\n74e7df6233 FIP \u0026 QoS: Check if QoS Policy is visible for user\nf502deee9c bgp: Do not learn routes from main BGP LRP\n05db9a2d55 DHCP: Clean HAProxy processes from ns during port_delete\ne161372955 Use RBAC API policy to create default security group\nf8c429d589 ml2: notify mechanism drivers in reverse order during port delete\n0ec8c22153 bgp: Redistribute local only addresses\n3f1ce7f950 [CI][FT] Tempoarary skip failing tests centos 9 stream\n794cfbe187 [iptables_hybrid] Fix conntrack zone not being restored on restart\n0e7eed7fcf Allow shrinking a segment range to an allocated boundary\n017469463f ovn: Don\u0027t wipe network HA_Chassis_Group on empty Gateway_Chassis\nda18231f98 Do not update unmodified ACLs in logapi service\nc7d0dc15bf Fix accidental removal of router external gateway info\n70d1c4e72f Validate RBAC ``target_project`` as UUID or wildcard\n94b8d4b1d4 Switch from setting LSP.tag to LSP.tag_request\n186c1a18ca Prevent cross-project subnet onboard on shared networks\n6e325eff58 Reject overlapping network segment ranges for tunnel types\nfcd54c1183 quota: Fix reservation under-counting in ``get_total_reservations_map()``\n0b01b68f1d qos: Validate parent policy ownership in ``get_policy_rule``\n0c1b467798 ovn: Ignore network HCGs in ``HAChassisGroupRouterEvent``\n1bfdccc2ce bgp: Create new InterconnectBridge\nfe5a333412 Fix load_db_migration_drivers\n5ebbea7055 ovn: Skip LSP host info update for router ports\n6479d51d6b ovn: read LSP ``up`` state from IDL cache in ``update_lsp_host_info``\nbba77b6ece rpc/ovn: Use base OVN IDL for RPC workers, disable notification events\n2bc474a7db [functional fips] Switch venv prep to python3.12\n408d683d71 segments: Catch RP not found in ``_delete_nova_inventory``\n3f0e8a3327 ovn: Fix HA_Chassis_Group race condition in port create/update\nb234755092 [UT] Update base_mac config override\ncae08d7db8 bgp: Limit lrp_mac to the chassis\n\nChange-Id: I63d6ecb7a13faaba2faef039776365424ea6f826\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1eb79e00770b9c24dec3e5dbc696f0ff24fc507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1eb79e00770b9c24dec3e5dbc696f0ff24fc507"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
