)]}'
{"id":"openstack%2Freleases~1005721","triplet_id":"openstack%2Freleases~master~I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce","project":"openstack/releases","branch":"master","topic":"indri-init-series","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-22 09:24:53.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce","subject":"Add 2027.1 Indri deliverable files (library, client-library)","status":"NEW","created":"2026-09-15 13:04:14.000000000","updated":"2026-10-02 16:35:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf139807bb3b50c14afb146a8b6436f16a41815a","_number":1005721,"virtual_id_number":1005721,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-02 16:02:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2026-10-02 15:41:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-10-02 16:35:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-10-02 15:41:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2026-10-02 15:41:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-10-02 15:41:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-10-02 15:41:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-10-02 16:35:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 13:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 13:24:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 09:24:53.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-09-28 12:23:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-09-28 14:18:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-28 14:18:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-10-02 10:31:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-10-02 16:35:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"684cc8456b4018441637c54b22c153d7d2cdd94f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 13:04:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0de1c4a95ab179a86575cd3f2ea9ed029c155f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 13:24:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e672b5cade8d40eba6087e547b6a1d27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abe93547fc74f53b44e274744524e5e : SUCCESS in 7m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3338cf579a594d75994147c96f3fdfcb : SUCCESS in 6m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d9e2842837624a5d856633490a051a52 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3a99ee404b6ccb638fd9dc13bb79139b5ff7ed45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 14:25:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"354eefb2c879a2cb550edd8307e700a1a9355b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:41:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a402bb4d4747fc948e653cf02684d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2708a1c38fc430fabe7553b3764e002 : SUCCESS in 4m 51s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/54d381705e524eff9af0d12d78ca8b72 : SUCCESS in 9m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/20b3691be5fa451faa7aaef2ede87d43 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d7c45d91a834dbaaaaed99fd80c471165df8b556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:28:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f64edaf3a7d757797970308b620e08f96cf023bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:54:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5589f517edb453caa6f9837e2e79ecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51870472ba846fdb719ecd2d2772ccd : SUCCESS in 4m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/46367ff327544a2abbde945a7c89a4c3 : SUCCESS in 10m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/711c7e11403c41e19c2a1b44ec17c87a : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b9e37efebda1877a0d08c7e24ee08e97d9735472","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-22 09:24:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9811be79cf0ce6aee0abb3ba627d28dc497becf9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-09-28 12:23:59.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"332e7a322fbb2d196762f89eaa56d21f89ed0558","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-28 14:18:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c346a0b09dcb4bbf9a1071c97ddba0078f3d51e8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-28 14:18:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85d044187b9553b71b4cd510cf299a79a2e5aefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:19:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"27b85af5698aee1d1197c18e4b7e74c825dcdc36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:20:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"a5d21557cad1dc9516039caa440112ae5723a2fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:21:45.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"56abd6b758a0502012a668852084a79ad44688f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:30:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2e0567386c4e2fbc5cbb4ba2bb213f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b3a1cfe7ed140bf934d4bfcb3022ddd : SUCCESS in 4m 37s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/2a3b02a0cef14759809300755c9f4b8b : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5f442257df8d4b149c4b279b43060d6c : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6},{"id":"8e6110876b932f8fdae6eb7e2fa38d3927c76fe2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 13:10:44.000000000","message":"Topic set to indri-init-series","accounts_in_message":[],"_revision_number":6},{"id":"91a658ba69b6a1d69063fa33097f57682ec93203","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-02 10:31:20.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"066b3800deba92382b68bd8de4385102d1015bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 15:41:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"cb4a4703273d9700055bcb60c8a52bcb8a1ad6ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:02:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f21a2d25d847aa86626c7d5c3ab5ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248e288b132549919ee887ea6b552e71 : SUCCESS in 7m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3b53105448384060befd6256808be343 : SUCCESS in 7m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f4c6f5a53b2c4fc0a3ce06ea50e54e20 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":7},{"id":"bf139807bb3b50c14afb146a8b6436f16a41815a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-10-02 16:35:21.000000000","message":"Patch Set 7: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479","revisions":{"0d287e09f0ec6add3e0abd2e4e05b1876e899b59":{"kind":"REWORK","_number":1,"created":"2026-09-15 13:04:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/1"}}},"commit":{"parents":[{"commit":"cb34ba45f817001f73a62286c2cabd3fb03f4f74","subject":"init-series: Add ability to filter by type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb34ba45f817001f73a62286c2cabd3fb03f4f74"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:04:00.000000000","tz":60},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d287e09f0ec6add3e0abd2e4e05b1876e899b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d287e09f0ec6add3e0abd2e4e05b1876e899b59"}]},"branch":"refs/heads/master"},"4c4210f6cf0e45516385961b7d6ba80b5cdb0a23":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-15 14:25:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/2"}}},"commit":{"parents":[{"commit":"1077a341cda7e5e197c7218719129b9ecdef7d2b","subject":"init-series: Add ability to filter by type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1077a341cda7e5e197c7218719129b9ecdef7d2b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 14:25:25.000000000","tz":60},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4210f6cf0e45516385961b7d6ba80b5cdb0a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4210f6cf0e45516385961b7d6ba80b5cdb0a23"}]},"branch":"refs/heads/master"},"5f709432fd9d6b71c910de34a40afa347c586792":{"kind":"NO_CHANGE","_number":3,"created":"2026-09-21 11:28:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/3"}}},"commit":{"parents":[{"commit":"b057de59134bc0dc473cfdceec77cc22f113959b","subject":"init-series: Add ability to filter by type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b057de59134bc0dc473cfdceec77cc22f113959b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:28:25.000000000","tz":0},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f709432fd9d6b71c910de34a40afa347c586792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f709432fd9d6b71c910de34a40afa347c586792"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1077a341cda7e5e197c7218719129b9ecdef7d2b","ours":"4c4210f6cf0e45516385961b7d6ba80b5cdb0a23","theirs":"b057de59134bc0dc473cfdceec77cc22f113959b","merge_strategy":"recursive","contains_conflicts":false}},"58f62b7ecb013d9430670ca300d07c1fcdb91fb5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-10-01 11:19:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/4"}}},"commit":{"parents":[{"commit":"2b98c7a9f62907f6b23bfe661110f0a2449b7c65","subject":"init-series: Add ability to filter by type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b98c7a9f62907f6b23bfe661110f0a2449b7c65"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 11:19:17.000000000","tz":60},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58f62b7ecb013d9430670ca300d07c1fcdb91fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58f62b7ecb013d9430670ca300d07c1fcdb91fb5"}]},"branch":"refs/heads/master"},"7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-10-01 11:20:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/5"}}},"commit":{"parents":[{"commit":"fa1510ece063d872505a76305c0f501aa206b311","subject":"init-series: Add ability to filter by type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa1510ece063d872505a76305c0f501aa206b311"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 11:19:46.000000000","tz":60},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c"}]},"branch":"refs/heads/master"},"4425b218c8a0c54f769beb91b0943e92fa05752c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-10-01 11:21:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/6"}}},"commit":{"parents":[{"commit":"73bd343b93b757f8e46d8dbf052c3063f82535d8","subject":"init-series: Add ability to filter by type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/73bd343b93b757f8e46d8dbf052c3063f82535d8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 11:21:45.000000000","tz":0},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4425b218c8a0c54f769beb91b0943e92fa05752c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4425b218c8a0c54f769beb91b0943e92fa05752c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fa1510ece063d872505a76305c0f501aa206b311","ours":"7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c","theirs":"73bd343b93b757f8e46d8dbf052c3063f82535d8","merge_strategy":"recursive","contains_conflicts":false}},"1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-10-02 15:41:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/21/1005721/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/1005721/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/1005721/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/1005721/7"}}},"commit":{"parents":[{"commit":"adcc7a5542784abc7a86b73cd5bbb33006a4761b","subject":"Merge \"[security] Release oslo.messaging 17.1.1, 17.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adcc7a5542784abc7a86b73cd5bbb33006a4761b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:02:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 15:41:33.000000000","tz":60},"subject":"Add 2027.1 Indri deliverable files (library, client-library)","message":"Add 2027.1 Indri deliverable files (library, client-library)\n\nThe generated content initializes the 2027.1 Indri cycle\u0027s deliverable\nfiles, based on 2026.2 Hibiscus deliverables.\n\nChange-Id: I6fd24960c397495eea6bcbd07ab46f7ac3e5c7ce\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nGenerated-By: openstack/releases:init_series --type library --type client-library\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY"},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
