)]}'
{"id":"openstack%2Freleases~1005909","triplet_id":"openstack%2Freleases~master~I0c3294c67f8a13c61bd018365ea1ae7c89e8056b","project":"openstack/releases","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-16 15:33:48.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0c3294c67f8a13c61bd018365ea1ae7c89e8056b","subject":"openstacksdk 4.21.0","status":"NEW","created":"2026-09-16 15:07:13.000000000","updated":"2026-09-21 11:41:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97e6d7056e8ad4d7a280f66aff3593e0c92f35b7","_number":1005909,"virtual_id_number":1005909,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-21 11:41:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-09-21 11:30:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 15:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d811f9f032ad120c90c45c3b9882b1b4d0f38bdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 15:07:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e23dd3dfb26c84d563cf16e50fd85cbd5b9ac7e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:11:21.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2963651ab66073ab5d88b27bbde223be5fb418ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:33:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d8f8e13f3d040a09dc26204bef727bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9f3568b15a4635b641b84d526869e4 : SUCCESS in 7m 38s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/82eeefa420f84b058b08df6337fc00a7 : FAILURE in 5m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d6becc06f35c4850a89e18d2eae892a9 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fb45036ef8b3f027137b8c215852adb72eb0e7d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:28:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"31114c0632d8697e2539aafd79026ada945f287d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:30:21.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"97e6d7056e8ad4d7a280f66aff3593e0c92f35b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:41:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e882c9e2bd374e52a47d3047a71f2892\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5b43a7df184ea3b3c1784620eb9bf0 : SUCCESS in 9m 02s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e566bf821e554e2e95f1c047314bcd43 : FAILURE in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/14a12c352a2d4b4fa4a3597ff82d4e30 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c3953470a6e99e577388f01b17a4773a851a2c2","revisions":{"fbd5a81ff79b5ff3233bb862c5318f1cc46379e3":{"kind":"REWORK","_number":1,"created":"2026-09-16 15:07:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/1005909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/1005909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/1005909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/1005909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/1005909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/1005909/1"}}},"commit":{"parents":[{"commit":"4c4210f6cf0e45516385961b7d6ba80b5cdb0a23","subject":"Add 2027.1 Indri deliverable files (library, client-library)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c4210f6cf0e45516385961b7d6ba80b5cdb0a23"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:02:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:02:17.000000000","tz":60},"subject":"openstacksdk 4.21.0","message":"openstacksdk 4.21.0\n\nNote that the removal of TagMixin from Domain was a bugfix: this never\nworked so we can\u0027t break anyone.\n\n❯ git log 4.20.0..b22ad455f --oneline --no-decorate --no-merges\nb22ad455f typing: Remove unnecessary cast\n6ee4020b5 zuul: Make openstacksdk-functional-devstack-magnum voting\n7189458a3 functional: Reorganise cloud functional tests by service\n45e4c4318 key-manager: Fix SecretACL proxy and broken functional tests\n5b3e84fa6 docs: Remove transition_from_profile doc\n55def8e1e resource: Fix re-entrant alias resolution\n5d30925a1 config: Separate parsing of .cache section\n2e5f603ec docs: Update metrics guide\naf74f7bd3 typing: Simplify QueryParamters type handling further\nf6fb581d3 typing: Simplify QueryParamters type handling\n1c50568d2 typing: Resolve TODO\n6a9248628 Ignore AuthPluginException in get_all()\n7e183238d image: Add support for cache clean and prune APIs\nf9fe33bcc Fix typos\n0c49994ac fix: Close underlying keystoneauth connection on exit\nd07f35852 stats: Fix prometheus histogram units\ne251f957d identity: Remove TagMixin from Domain\n549a741d6 config: Separate parsing of .metrics section\n659737b71 config: Separate parsing of .client section\nb207acd9c config: Add missing docs for CloudRegion\nedfce0de2 config: Rewrap exception message\n1d857a09d docs: Move MFA section under auth\n697f555e7 Update master for stable/2026.2\n4830e2f91 Fix typo\n\nChange-Id: I0c3294c67f8a13c61bd018365ea1ae7c89e8056b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbd5a81ff79b5ff3233bb862c5318f1cc46379e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbd5a81ff79b5ff3233bb862c5318f1cc46379e3"}]},"branch":"refs/heads/master"},"2c3953470a6e99e577388f01b17a4773a851a2c2":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-21 11:28:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/1005909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/09/1005909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/1005909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/1005909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/09/1005909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/09/1005909/2"}}},"commit":{"parents":[{"commit":"5f709432fd9d6b71c910de34a40afa347c586792","subject":"Add 2027.1 Indri deliverable files (library, client-library)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f709432fd9d6b71c910de34a40afa347c586792"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:02:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:28:36.000000000","tz":0},"subject":"openstacksdk 4.21.0","message":"openstacksdk 4.21.0\n\nNote that the removal of TagMixin from Domain was a bugfix: this never\nworked so we can\u0027t break anyone.\n\n❯ git log 4.20.0..b22ad455f --oneline --no-decorate --no-merges\nb22ad455f typing: Remove unnecessary cast\n6ee4020b5 zuul: Make openstacksdk-functional-devstack-magnum voting\n7189458a3 functional: Reorganise cloud functional tests by service\n45e4c4318 key-manager: Fix SecretACL proxy and broken functional tests\n5b3e84fa6 docs: Remove transition_from_profile doc\n55def8e1e resource: Fix re-entrant alias resolution\n5d30925a1 config: Separate parsing of .cache section\n2e5f603ec docs: Update metrics guide\naf74f7bd3 typing: Simplify QueryParamters type handling further\nf6fb581d3 typing: Simplify QueryParamters type handling\n1c50568d2 typing: Resolve TODO\n6a9248628 Ignore AuthPluginException in get_all()\n7e183238d image: Add support for cache clean and prune APIs\nf9fe33bcc Fix typos\n0c49994ac fix: Close underlying keystoneauth connection on exit\nd07f35852 stats: Fix prometheus histogram units\ne251f957d identity: Remove TagMixin from Domain\n549a741d6 config: Separate parsing of .metrics section\n659737b71 config: Separate parsing of .client section\nb207acd9c config: Add missing docs for CloudRegion\nedfce0de2 config: Rewrap exception message\n1d857a09d docs: Move MFA section under auth\n697f555e7 Update master for stable/2026.2\n4830e2f91 Fix typo\n\nChange-Id: I0c3294c67f8a13c61bd018365ea1ae7c89e8056b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c3953470a6e99e577388f01b17a4773a851a2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c3953470a6e99e577388f01b17a4773a851a2c2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4c4210f6cf0e45516385961b7d6ba80b5cdb0a23","ours":"fbd5a81ff79b5ff3233bb862c5318f1cc46379e3","theirs":"5f709432fd9d6b71c910de34a40afa347c586792","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
