)]}'
{"id":"openstack%2Freleases~1005911","triplet_id":"openstack%2Freleases~master~I280cbd599e0616e82af8d077fa23738889422432","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-10-09 07:18:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I280cbd599e0616e82af8d077fa23738889422432","subject":"python-openstackclient 10.4.0","status":"MERGED","created":"2026-09-16 15:07:13.000000000","updated":"2026-10-09 07:18:01.000000000","submitted":"2026-10-09 07:18:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1005911","meta_rev_id":"ce7f7909d9fade7b1082b2eb21b2b23c7bfc0db9","_number":1005911,"virtual_id_number":1005911,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-09 07:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-10-09 07:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-10-09 07:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-10-08 11:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 15:11:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-09 07:06:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"aabc9e97c81c6cc40536a46805eae8cbb357f013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 15:07:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f62681f4f97e59f7232f5f05d16a0bc045d86647","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:11:30.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"903e6eddc844b065c526b109b7f825e4e80cc814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:33:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6131242cbde4d3a9e65b4c4662acc7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8c687497e04d4e8df1a40493e1d018 : SUCCESS in 7m 23s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4b6796c19a8940a1adb8c4773ae32768 : FAILURE in 7m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/173e74802185444c823370ed376e60e0 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c813787f27dac48f9002dcbafc072f3705346d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:28:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c55553f80e0df6a02ebad80e85a0d02aa095a976","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:30:22.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"622adcb8fb2318888eed7e5365367cd6a8c2f3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:35:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b5edb66077e4d3c950c1217e65fbacf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c07b73f34c43a7ae04cbd5ffec97db : SUCCESS in 4m 37s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/43484fef35d4485fb375a994b7111114 : FAILURE in 4m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/395dfb58a603434e92968907625694d1 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ee326fd581e662514d05bffccb587dd3f6904ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:20:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a0f8166bb143bdd8bab7415eda006e8381fd1b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:22:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"9c6d8d246f2a56a0ab43286a8b5763b66f5c29bf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:24:39.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"b8da60bd317b6117b39074afa909f5144249eb27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:29:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19061e23a324db6858e418394930710\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e959f0a57740d5a69ecb8cf07cfbeb : SUCCESS in 4m 49s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1a2f642f24014773a8c57f26fee452a8 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/03c9ba1938424837b34b27b85184e2f5 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":4},{"id":"c985657d71c4b13e1e4118b072dede641d91316b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 15:42:07.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5e9aec592ffb9d0ee38b05dee8d33324236f905c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 15:43:46.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"10550f91a485de0ea50a6abf1a4ffc41100aeb24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:07:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cebbe1f8f434dbf8216eb78947a671a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d4b7f2158334651baac0b64f25903b6 : SUCCESS in 8m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/88f5d997f92d424a903f6b8de020a5ac : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7c5c111c4a864aa7885b275e9e8e5ceb : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"7e57ee97cd140787996a9c9f658411a47f54ca88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-08 11:12:33.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"78078edb8ff7739d62420fe05b35314e5983c2d1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-08 11:13:31.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"9fd42bcf937dca1a65baf33e0fd634b55b9bc328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-08 11:29:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/174cef0338254dda8525cab9698f14aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f99628dca94551ac6fb7f8578e5984 : SUCCESS in 6m 36s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/96a2839ffa254d02aaf045c199ab6768 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7963db511d1a4c88b3dc89028f15cfe9 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":6},{"id":"b7800a6b95b902f9a0f9ec99efb0239a3f3b0037","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-10-09 07:06:44.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cd5c540cc7fb0e48c5d2b8bd65acd2ce721591e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 07:06:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cc76b4b3fbe6f6322221bea6c55d351186239f91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 07:18:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2514060410104f89980e36c95b8b9ba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06509e40a7624110b8c4961e2723cea3 : SUCCESS in 5m 27s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c84e8b04c3d14aeda9dfe2f0ea81449e : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":6},{"id":"ce7f7909d9fade7b1082b2eb21b2b23c7bfc0db9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-09 07:18:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7ea49472c535a1465ce785dd92efb6b08c2708e3","revisions":{"e7a60038d4f5f5d1bcf3a94af28c9a646e65b4b5":{"kind":"REWORK","_number":1,"created":"2026-09-16 15:07:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/1005911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/1005911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/1005911/1"}}},"commit":{"parents":[{"commit":"7c14126d1f25733ae2a3cc22014cd9d24a37cc86","subject":"keystoneauth 5.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c14126d1f25733ae2a3cc22014cd9d24a37cc86"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"subject":"python-openstackclient 10.4.0","message":"python-openstackclient 10.4.0\n\n❯ git log 10.3.0..47b119656 --oneline --no-decorate --no-merges\n839c132db vpnaas: Remove deprecated sha1 and 3des algorithms\nf5cf1fb30 Migrate \u0027endpoint group\u0027 commands to SDK\n2c2df481e image: add --store-preference to \"image save\"\n6ef4d932a Update master for stable/2026.2\n7cf74120c Network: Add onboard_network_subnets\n6d46ca8d4 bump minimum version of openstacksdk\n08dd55889 Network: Simplify ip_availability_details access\nb660824b8 Add a \"cloud cache delete\" command\n\nChange-Id: I280cbd599e0616e82af8d077fa23738889422432\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7a60038d4f5f5d1bcf3a94af28c9a646e65b4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7a60038d4f5f5d1bcf3a94af28c9a646e65b4b5"}]},"branch":"refs/heads/master"},"7d70e86f255369377dc01a214147981cbde04d03":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-21 11:28:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/1005911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/1005911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/1005911/2"}}},"commit":{"parents":[{"commit":"a5d194c27e653deea3d70866b8d9eb3060be6615","subject":"keystoneauth 5.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5d194c27e653deea3d70866b8d9eb3060be6615"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:28:45.000000000","tz":0},"subject":"python-openstackclient 10.4.0","message":"python-openstackclient 10.4.0\n\n❯ git log 10.3.0..47b119656 --oneline --no-decorate --no-merges\n839c132db vpnaas: Remove deprecated sha1 and 3des algorithms\nf5cf1fb30 Migrate \u0027endpoint group\u0027 commands to SDK\n2c2df481e image: add --store-preference to \"image save\"\n6ef4d932a Update master for stable/2026.2\n7cf74120c Network: Add onboard_network_subnets\n6d46ca8d4 bump minimum version of openstacksdk\n08dd55889 Network: Simplify ip_availability_details access\nb660824b8 Add a \"cloud cache delete\" command\n\nChange-Id: I280cbd599e0616e82af8d077fa23738889422432\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d70e86f255369377dc01a214147981cbde04d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7d70e86f255369377dc01a214147981cbde04d03"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7c14126d1f25733ae2a3cc22014cd9d24a37cc86","ours":"e7a60038d4f5f5d1bcf3a94af28c9a646e65b4b5","theirs":"a5d194c27e653deea3d70866b8d9eb3060be6615","merge_strategy":"recursive","contains_conflicts":false}},"3534822434a37b50c3095f787caec8c82ca52723":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-10-01 11:20:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/1005911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/1005911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/1005911/3"}}},"commit":{"parents":[{"commit":"14ab043515c9e9cd37a7e67b618f08753ddb2f7e","subject":"keystoneauth 5.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14ab043515c9e9cd37a7e67b618f08753ddb2f7e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 11:20:37.000000000","tz":0},"subject":"python-openstackclient 10.4.0","message":"python-openstackclient 10.4.0\n\n❯ git log 10.3.0..47b119656 --oneline --no-decorate --no-merges\n839c132db vpnaas: Remove deprecated sha1 and 3des algorithms\nf5cf1fb30 Migrate \u0027endpoint group\u0027 commands to SDK\n2c2df481e image: add --store-preference to \"image save\"\n6ef4d932a Update master for stable/2026.2\n7cf74120c Network: Add onboard_network_subnets\n6d46ca8d4 bump minimum version of openstacksdk\n08dd55889 Network: Simplify ip_availability_details access\nb660824b8 Add a \"cloud cache delete\" command\n\nChange-Id: I280cbd599e0616e82af8d077fa23738889422432\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3534822434a37b50c3095f787caec8c82ca52723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3534822434a37b50c3095f787caec8c82ca52723"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a5d194c27e653deea3d70866b8d9eb3060be6615","ours":"7d70e86f255369377dc01a214147981cbde04d03","theirs":"14ab043515c9e9cd37a7e67b618f08753ddb2f7e","merge_strategy":"recursive","contains_conflicts":false}},"a45c38ef80e0d038f01e835624f3c77514a820c4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-10-01 11:22:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/1005911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/1005911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/1005911/4"}}},"commit":{"parents":[{"commit":"75b24553307d960e34fb49ec4c028ddd88bc1ef1","subject":"keystoneauth 5.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75b24553307d960e34fb49ec4c028ddd88bc1ef1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-01 11:22:17.000000000","tz":0},"subject":"python-openstackclient 10.4.0","message":"python-openstackclient 10.4.0\n\n❯ git log 10.3.0..47b119656 --oneline --no-decorate --no-merges\n839c132db vpnaas: Remove deprecated sha1 and 3des algorithms\nf5cf1fb30 Migrate \u0027endpoint group\u0027 commands to SDK\n2c2df481e image: add --store-preference to \"image save\"\n6ef4d932a Update master for stable/2026.2\n7cf74120c Network: Add onboard_network_subnets\n6d46ca8d4 bump minimum version of openstacksdk\n08dd55889 Network: Simplify ip_availability_details access\nb660824b8 Add a \"cloud cache delete\" command\n\nChange-Id: I280cbd599e0616e82af8d077fa23738889422432\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a45c38ef80e0d038f01e835624f3c77514a820c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a45c38ef80e0d038f01e835624f3c77514a820c4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"14ab043515c9e9cd37a7e67b618f08753ddb2f7e","ours":"3534822434a37b50c3095f787caec8c82ca52723","theirs":"75b24553307d960e34fb49ec4c028ddd88bc1ef1","merge_strategy":"recursive","contains_conflicts":false}},"6954a7f66eae7c55c5be46d6f52968444a788766":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-10-02 15:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/1005911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/1005911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/1005911/5"}}},"commit":{"parents":[{"commit":"fc5a604818fd5d1a2246f03b79980be8966c1ff2","subject":"keystoneauth 5.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc5a604818fd5d1a2246f03b79980be8966c1ff2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 15:42:07.000000000","tz":0},"subject":"python-openstackclient 10.4.0","message":"python-openstackclient 10.4.0\n\n❯ git log 10.3.0..47b119656 --oneline --no-decorate --no-merges\n839c132db vpnaas: Remove deprecated sha1 and 3des algorithms\nf5cf1fb30 Migrate \u0027endpoint group\u0027 commands to SDK\n2c2df481e image: add --store-preference to \"image save\"\n6ef4d932a Update master for stable/2026.2\n7cf74120c Network: Add onboard_network_subnets\n6d46ca8d4 bump minimum version of openstacksdk\n08dd55889 Network: Simplify ip_availability_details access\nb660824b8 Add a \"cloud cache delete\" command\n\nChange-Id: I280cbd599e0616e82af8d077fa23738889422432\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6954a7f66eae7c55c5be46d6f52968444a788766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6954a7f66eae7c55c5be46d6f52968444a788766"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"75b24553307d960e34fb49ec4c028ddd88bc1ef1","ours":"a45c38ef80e0d038f01e835624f3c77514a820c4","theirs":"fc5a604818fd5d1a2246f03b79980be8966c1ff2","merge_strategy":"recursive","contains_conflicts":false}},"7ea49472c535a1465ce785dd92efb6b08c2708e3":{"kind":"REWORK","_number":6,"created":"2026-10-08 11:12:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/1005911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/1005911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/1005911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/1005911/6"}}},"commit":{"parents":[{"commit":"56b4b706716b8946ac9cdb1a96c794a6050093cf","subject":"Merge \"[tool] Fix validate_series_open\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56b4b706716b8946ac9cdb1a96c794a6050093cf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 15:06:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-08 11:11:27.000000000","tz":60},"subject":"python-openstackclient 10.4.0","message":"python-openstackclient 10.4.0\n\nNote the removal of deprecated sha1 and 3des algorithms for the vpnaas\ncommands (commit 839c132db) was partially reverted in commit 9e4e4ecf5\nto avoid a major version bump.\n\nChange-Id: I280cbd599e0616e82af8d077fa23738889422432\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ea49472c535a1465ce785dd92efb6b08c2708e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ea49472c535a1465ce785dd92efb6b08c2708e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
