)]}'
{"id":"openstack%2Freleases~1006695","triplet_id":"openstack%2Freleases~master~I26472a549c8f0b97e63071009b30877c62b07af9","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-10-06 14:39:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26472a549c8f0b97e63071009b30877c62b07af9","subject":"Release sushy 5.14.0","status":"MERGED","created":"2026-09-22 09:27:51.000000000","updated":"2026-10-06 14:39:40.000000000","submitted":"2026-10-06 14:39:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1006695","meta_rev_id":"10d6305ae016ad93a835d9fef4dd9466d4ac825e","_number":1006695,"virtual_id_number":1006695,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-06 14:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-10-06 14:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-10-06 07:25:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-10-06 14:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2026-10-06 07:26:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 09:29:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 01:34:04.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2026-10-06 14:12:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-10-06 14:26:01.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"897d68bf390bd17c5a52e2d232c22c5008c2efd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-09-22 09:27:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d4b6b16e6ab526b504bb54da5f860a299251c5f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:29:23.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2f82db7d986f797c3c2b40ca7c1694c12eaf0085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 09:42:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cab6c6466364d41ad6a13cf4fef7771\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1befa8298a464b1d903ee92021adc90c : SUCCESS in 6m 48s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7b45ab95331b452eb35eb378f7362271 : FAILURE in 3m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ce9bcca17b2c4809b6926baa8f7c63d6 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d2f07fc406f80435f406c6168cb53b7b9cbf7bf6","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-09-23 01:34:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e39066fefe8841db2ce683983a043a17b3ae61b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:20:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_23851\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"_revision_number":2},{"id":"0ffe12f5f0077525c1f2e2755e532f39014b837c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-01 11:22:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_23851\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n","accounts_in_message":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"_revision_number":3},{"id":"e7cf0c24707496a15aea0518ffecd424621b1b88","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:24:31.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"ebc9633910f03f94de1bab394f686d5ba55f0d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:33:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70f1baee34f44775b26c950ef0b9fcd5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9585db9c975342a99ccb1083a884ef90 : SUCCESS in 7m 14s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6cdcd397aebf4fc6810f2ecce5f9e97d : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a25e29d3415541c3b2bcc68d3f00e9ec : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e16e17aaf264953651f281523255f34cfdba1c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 15:42:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_23851\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"_revision_number":4},{"id":"ebfefddd5055350ba03c8412b77419a55df4bfb4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 15:43:41.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"72e0119516f87f084e59a9f0bc18ef080032dddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:07:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be73ebf4502a4818af75ec9ea3fcf0d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90bce57130740c8aeaa0231397d4c7c : SUCCESS in 8m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/633191660803451bac47e1e351c61b0c : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9725dce508ff4b4e80417b2bc9463660 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":4},{"id":"680184c1237a036ab976117741bca995739a96c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-10-06 07:25:46.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"af923b1e889324dedeb46536339869555dac765a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 07:26:33.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"d9a202dc4f2cad61c949529199da93c8c82368a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 07:35:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3c1a23a680460e8da45adcfd164742\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4c677f351349228a694127c7d5b29a : SUCCESS in 6m 39s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/8eaec6b3cb5b4ca88301c730ea8d3d65 : SUCCESS in 5m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a18ada92a5cc459f950dc861130a8dbc : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":5},{"id":"833ace0942ea15ab049e1ee5ab76e6305a9fdb7e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-10-06 14:12:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0df28d5749e1815774b5ac894c6d1f90dc428db3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-10-06 14:26:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"76ad09cf07e9230fabd889e13df21635580ba465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 14:26:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a9bbfb0d255f098466b051b8df524dbbe4820d55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 14:39:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c5a73702aa4e43848c1dd7fc43b0ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cba842d80914287b06967f03126b003 : SUCCESS in 7m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ddfc9f2857674d998fcced0abbdb9a85 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":5},{"id":"10d6305ae016ad93a835d9fef4dd9466d4ac825e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 14:39:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e163c44d80ef1fd42a269136e75d36f4dce7e175","revisions":{"6a0da989087b60da1670ee0309aed8c3f83bc5bf":{"kind":"REWORK","_number":1,"created":"2026-09-22 09:27:51.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/95/1006695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/1006695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/1006695/1"}}},"commit":{"parents":[{"commit":"5f709432fd9d6b71c910de34a40afa347c586792","subject":"Add 2027.1 Indri deliverable files (library, client-library)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f709432fd9d6b71c910de34a40afa347c586792"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-09-22 09:27:13.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-09-22 09:27:42.000000000","tz":120},"subject":"Release sushy 5.14.0","message":"Release sushy 5.14.0\n\nChange-Id: I26472a549c8f0b97e63071009b30877c62b07af9\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a0da989087b60da1670ee0309aed8c3f83bc5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a0da989087b60da1670ee0309aed8c3f83bc5bf"}]},"branch":"refs/heads/master"},"9d7bc2387e01756390e6055c2bf4fe507187a874":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-10-01 11:20:51.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/1006695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/1006695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/1006695/2"}}},"commit":{"parents":[{"commit":"7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c","subject":"Add 2027.1 Indri deliverable files (library, client-library)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-09-22 09:27:13.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-10-01 11:20:51.000000000","tz":0},"subject":"Release sushy 5.14.0","message":"Release sushy 5.14.0\n\nChange-Id: I26472a549c8f0b97e63071009b30877c62b07af9\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d7bc2387e01756390e6055c2bf4fe507187a874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d7bc2387e01756390e6055c2bf4fe507187a874"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5f709432fd9d6b71c910de34a40afa347c586792","ours":"6a0da989087b60da1670ee0309aed8c3f83bc5bf","theirs":"7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c","merge_strategy":"recursive","contains_conflicts":false}},"bcc40bc0b34c778fe5fdf7f081435e41e462f89a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-10-01 11:22:31.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/1006695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/1006695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/1006695/3"}}},"commit":{"parents":[{"commit":"4425b218c8a0c54f769beb91b0943e92fa05752c","subject":"Add 2027.1 Indri deliverable files (library, client-library)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4425b218c8a0c54f769beb91b0943e92fa05752c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-09-22 09:27:13.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-10-01 11:22:31.000000000","tz":0},"subject":"Release sushy 5.14.0","message":"Release sushy 5.14.0\n\nChange-Id: I26472a549c8f0b97e63071009b30877c62b07af9\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcc40bc0b34c778fe5fdf7f081435e41e462f89a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcc40bc0b34c778fe5fdf7f081435e41e462f89a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7c1f85fcc93de5001a8fc7ea83ca62a2c365f33c","ours":"9d7bc2387e01756390e6055c2bf4fe507187a874","theirs":"4425b218c8a0c54f769beb91b0943e92fa05752c","merge_strategy":"recursive","contains_conflicts":false}},"cad534f66342662f220e3f805d53df1ddb35dfed":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-10-02 15:42:17.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/1006695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/1006695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/1006695/4"}}},"commit":{"parents":[{"commit":"1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479","subject":"Add 2027.1 Indri deliverable files (library, client-library)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-09-22 09:27:13.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-10-02 15:42:17.000000000","tz":0},"subject":"Release sushy 5.14.0","message":"Release sushy 5.14.0\n\nChange-Id: I26472a549c8f0b97e63071009b30877c62b07af9\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cad534f66342662f220e3f805d53df1ddb35dfed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cad534f66342662f220e3f805d53df1ddb35dfed"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4425b218c8a0c54f769beb91b0943e92fa05752c","ours":"bcc40bc0b34c778fe5fdf7f081435e41e462f89a","theirs":"1b4d4eb53b2a23241fdb2bd6bd28a8d5f5a7b479","merge_strategy":"recursive","contains_conflicts":false}},"e163c44d80ef1fd42a269136e75d36f4dce7e175":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-10-06 07:25:46.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/95/1006695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/1006695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/1006695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/1006695/5"}}},"commit":{"parents":[{"commit":"48182b3c0f50d99776de662701daa4f8dcb36e8c","subject":"Test new signing key for 2027.1 Indri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/48182b3c0f50d99776de662701daa4f8dcb36e8c"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-09-22 09:27:13.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2026-10-06 07:25:34.000000000","tz":120},"subject":"Release sushy 5.14.0","message":"Release sushy 5.14.0\n\nChange-Id: I26472a549c8f0b97e63071009b30877c62b07af9\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e163c44d80ef1fd42a269136e75d36f4dce7e175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e163c44d80ef1fd42a269136e75d36f4dce7e175"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{"label:PTL-Approved\u003e0":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
