)]}'
{"id":"openstack%2Freleases~391715","triplet_id":"openstack%2Freleases~master~I38429dd29767afdbf395cbb3777879636ea0ad55","project":"openstack/releases","branch":"master","topic":"goal-remove-incubated-oslo-code","hashtags":[],"change_id":"I38429dd29767afdbf395cbb3777879636ea0ad55","subject":"Don\u0027t include openstack/common in flake8 exclude list","status":"MERGED","created":"2016-10-31 04:52:21.000000000","updated":"2016-11-01 19:04:53.000000000","submitted":"2016-11-01 19:04:53.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f888773a904fafc119292d3411363979babbeebc","_number":391715,"virtual_id_number":391715,"owner":{"_account_id":22255,"name":"Tuan","email":"tuanla@vn.fujitsu.com","username":"Tuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-11-01 19:04:53.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-11-01 18:59:28.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-11-01 18:59:28.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-11-01 18:59:28.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2016-11-01 19:04:53.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b8baf434c5099139d75e0eda6cc70bc149caa38a","author":{"_account_id":22255,"name":"Tuan","email":"tuanla@vn.fujitsu.com","username":"Tuan"},"date":"2016-10-31 04:52:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd6ec651c4322ac8236fc69e5a560efc7042bf5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-31 04:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-releases-tox-validate-ubuntu-xenial http://logs.openstack.org/15/391715/1/check/gate-releases-tox-validate-ubuntu-xenial/ac9dd76/ : SUCCESS in 2m 08s\n- gate-releases-tox-list-changes-ubuntu-xenial http://logs.openstack.org/15/391715/1/check/gate-releases-tox-list-changes-ubuntu-xenial/678bfdd/ : SUCCESS in 1m 16s\n- gate-releases-pep8-ubuntu-xenial http://logs.openstack.org/15/391715/1/check/gate-releases-pep8-ubuntu-xenial/4dd7fe6/ : SUCCESS in 1m 20s\n- gate-releases-docs-ubuntu-xenial http://docs-draft.openstack.org/15/391715/1/check/gate-releases-docs-ubuntu-xenial/b5d09b2//doc/build/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b39a566f8b8841d2bfee4be3bd2ce5fc4f0ec429","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2016-11-01 18:59:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18ca0442d7a0a0367899899aa7fe684b266e2721","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2016-11-01 19:00:31.000000000","message":"Topic changed from fixt to goal-remove-incubated-oslo-code","accounts_in_message":[],"_revision_number":1},{"id":"9bb8db3e43ff9ac97088eb2372e06fe2f7b0dd89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-01 19:01:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e66952944225e420c83ca04a4ba290902de7b530","date":"2016-11-01 19:04:53.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"f888773a904fafc119292d3411363979babbeebc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-01 19:04:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-releases-tox-validate-ubuntu-xenial http://logs.openstack.org/15/391715/1/gate/gate-releases-tox-validate-ubuntu-xenial/21aa7b0/ : SUCCESS in 1m 55s\n- gate-releases-docs-ubuntu-xenial http://docs-draft.openstack.org/15/391715/1/gate/gate-releases-docs-ubuntu-xenial/763d40f//doc/build/html/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbaf7f1412beadefa47a3702e5277429dcf669b6","revisions":{"fbaf7f1412beadefa47a3702e5277429dcf669b6":{"kind":"REWORK","_number":1,"created":"2016-10-31 04:52:21.000000000","uploader":{"_account_id":22255,"name":"Tuan","email":"tuanla@vn.fujitsu.com","username":"Tuan"},"ref":"refs/changes/15/391715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/391715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/391715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/391715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/391715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/391715/1"}}},"commit":{"parents":[{"commit":"a0613747f3e44a4e5f828da604bbf069b24fce4b","subject":"Merge \"Release OpenStack-Ansible Liberty/12.2.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0613747f3e44a4e5f828da604bbf069b24fce4b"}]}],"author":{"name":"Luong Anh Tuan","email":"tuanla@vn.fujitsu.com","date":"2016-10-31 04:49:20.000000000","tz":420},"committer":{"name":"Luong Anh Tuan","email":"tuanla@vn.fujitsu.com","date":"2016-10-31 04:51:49.000000000","tz":420},"subject":"Don\u0027t include openstack/common in flake8 exclude list","message":"Don\u0027t include openstack/common in flake8 exclude list\n\nThere is no this directory openstack/common which was used to keep codes from\noslo-incubator, we have retired oslo-incubator. Removing openstack/commonfrom\nall OpenStack code in favor of the Oslo libraries is a project wide goal\nfor the Ocata release. So don\u0027t use this directory any more. We should drop it\nfor improving searching efficiency.\n\nChange-Id: I38429dd29767afdbf395cbb3777879636ea0ad55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbaf7f1412beadefa47a3702e5277429dcf669b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbaf7f1412beadefa47a3702e5277429dcf669b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
