)]}'
{"id":"openstack%2Freleases~460739","triplet_id":"openstack%2Freleases~master~I92578f1f2749412e088e736ee677ab0d9cbcc7c2","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I92578f1f2749412e088e736ee677ab0d9cbcc7c2","subject":"Prepare a new TripleO Ocata release","status":"MERGED","created":"2017-04-27 20:35:49.000000000","updated":"2017-04-28 01:03:28.000000000","submitted":"2017-04-28 01:03:28.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b25af4253cb5289823605d660f29006d64cdd556","_number":460739,"virtual_id_number":460739,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2017-04-28 01:03:28.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":2,"date":"2017-04-28 00:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-04-27 21:03:43.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":1,"date":"2017-04-28 00:58:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-27 21:03:43.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2017-04-28 00:58:31.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2017-04-28 01:03:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"32ff6e94c32b25ec66daf713cfed8c3886216dce","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-04-27 20:35:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b96e544b126746368bb5afe9756ddcf9c3d03598","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-27 20:44:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-releases-python35 http://logs.openstack.org/39/460739/1/check/gate-releases-python35/2999337/ : SUCCESS in 1m 50s\n- gate-releases-tox-validate-ubuntu-xenial http://logs.openstack.org/39/460739/1/check/gate-releases-tox-validate-ubuntu-xenial/68cad2f/ : SUCCESS in 6m 45s\n- gate-releases-tox-list-changes-ubuntu-xenial http://logs.openstack.org/39/460739/1/check/gate-releases-tox-list-changes-ubuntu-xenial/8733fc2/ : SUCCESS in 2m 50s\n- gate-releases-pep8-ubuntu-xenial http://logs.openstack.org/39/460739/1/check/gate-releases-pep8-ubuntu-xenial/1f81f63/ : SUCCESS in 2m 11s\n- gate-releases-docs-ubuntu-xenial http://docs-draft.openstack.org/39/460739/1/check/gate-releases-docs-ubuntu-xenial/ac72aca//doc/build/html/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8d1248889ec32cb1da3af9106585329beecb31ff","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-04-27 21:03:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3a528fa598d9b560d1e180633250db2d005c551","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2017-04-28 00:58:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcc37efaea758d786933ca222b4d2cd0f6ba6ca4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-28 00:58:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"031ddec09b396a71fc51bc9d0bf580f83fc69d1e","date":"2017-04-28 01:03:28.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"b25af4253cb5289823605d660f29006d64cdd556","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-28 01:03:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-releases-python35 http://logs.openstack.org/39/460739/1/gate/gate-releases-python35/e37e5a5/ : SUCCESS in 1m 40s\n- gate-releases-tox-validate-ubuntu-xenial http://logs.openstack.org/39/460739/1/gate/gate-releases-tox-validate-ubuntu-xenial/3ecc6bc/ : SUCCESS in 4m 37s\n- gate-releases-docs-ubuntu-xenial http://docs-draft.openstack.org/39/460739/1/gate/gate-releases-docs-ubuntu-xenial/ca80f04//doc/build/html/ : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"647f20304d56718f0d277e3ce865984cbfef214e","revisions":{"647f20304d56718f0d277e3ce865984cbfef214e":{"kind":"REWORK","_number":1,"created":"2017-04-27 20:35:49.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/39/460739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/39/460739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/460739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/460739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/39/460739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/39/460739/1"}}},"commit":{"parents":[{"commit":"ddbc1d91da7ecaa47bdcf3d92a50666e42f424c3","subject":"Merge \"mitaka eol: release os-win 0.4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ddbc1d91da7ecaa47bdcf3d92a50666e42f424c3"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-04-27 20:35:22.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-04-27 20:35:47.000000000","tz":-240},"subject":"Prepare a new TripleO Ocata release","message":"Prepare a new TripleO Ocata release\n\nChange-Id: I92578f1f2749412e088e736ee677ab0d9cbcc7c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/647f20304d56718f0d277e3ce865984cbfef214e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/647f20304d56718f0d277e3ce865984cbfef214e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ddbc1d91da7ecaa47bdcf3d92a50666e42f424c3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
