)]}'
{"id":"openstack%2Freleases~546280","triplet_id":"openstack%2Freleases~master~Ie4dc2226ea0765919a9e6eb400983260094784a7","project":"openstack/releases","branch":"master","topic":"upstream_stable","hashtags":[],"change_id":"Ie4dc2226ea0765919a9e6eb400983260094784a7","subject":"support pypi-name in repository-settings","status":"MERGED","created":"2018-02-20 19:20:45.000000000","updated":"2018-02-21 22:06:49.000000000","submitted":"2018-02-21 22:06:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"546280-1519250809713-6d4740cf","meta_rev_id":"1537a123bf9c6826ada31f941012d7d86dfc0f09","_number":546280,"virtual_id_number":546280,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-02-21 22:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2018-02-21 21:24:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2018-02-21 21:24:18.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2018-02-21 21:44:25.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-20 20:01:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-02-21 05:04:52.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2018-02-21 22:06:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1dd99abbd7e382739577fdee4ec91f1abc05d7d9","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-02-20 19:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffa207a13ec280f0a3c698f3eda18e5b169b54cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-20 19:33:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/80/546280/1/check/openstack-tox-py35/36838bb/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/80/546280/1/check/openstack-tox-pep8/5feacbe/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/80/546280/1/check/openstack-tox-validate/82d3c50/ : SUCCESS in 2m 56s\n- releases-tox-list-changes http://logs.openstack.org/80/546280/1/check/releases-tox-list-changes/879ca4b/ : SUCCESS in 3m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/546280/1/check/build-openstack-sphinx-docs/a577330/html/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9a24d1484a4839f155e98dd7f61b1aea67fc0013","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-02-20 20:01:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cbed329e46cf0f2eb14e496e61dbac7236bda68c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-02-20 22:56:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ac1b8f1f0ea31c959b1c761bd0f9c4c93a291b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-20 23:07:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/80/546280/2/check/openstack-tox-py35/71a95ed/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/80/546280/2/check/openstack-tox-pep8/6bf83f3/ : SUCCESS in 2m 34s\n- openstack-tox-validate http://logs.openstack.org/80/546280/2/check/openstack-tox-validate/f5cd799/ : SUCCESS in 2m 42s\n- releases-tox-list-changes http://logs.openstack.org/80/546280/2/check/releases-tox-list-changes/3d5a6c4/ : SUCCESS in 3m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/546280/2/check/build-openstack-sphinx-docs/eb0dc91/html/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c64b18cb5031c1967a5aea81272c9970c2725a93","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-02-21 05:04:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4578c1c54941ccfae41340716ee5e19bb1a9239","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-02-21 14:13:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7c601e10e9d81d7372105e67baf6ed8b40d5c07","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-02-21 21:24:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b8edf1a742eb579b9318436c21692cb75afa6dc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 21:32:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/80/546280/3/check/openstack-tox-py35/ffa837a/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/80/546280/3/check/openstack-tox-pep8/373cd9a/ : SUCCESS in 2m 15s\n- openstack-tox-validate http://logs.openstack.org/80/546280/3/check/openstack-tox-validate/16151d6/ : SUCCESS in 2m 51s\n- releases-tox-list-changes http://logs.openstack.org/80/546280/3/check/releases-tox-list-changes/f293d88/ : SUCCESS in 3m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/546280/3/check/build-openstack-sphinx-docs/f03dd7d/html/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"b372754057a0faebca9ec73ad117277d4f16e547","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-02-21 21:44:25.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d66da096b39a0ecb9e09f9cd29b576b0c114a425","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 21:52:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"621fa57b326589a7344a9b5eca6856644c7a9a3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 22:06:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d08f3ae6f5569670ef2136f096a64d085d340a38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 22:06:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/80/546280/3/gate/openstack-tox-py35/d9d28b6/ : SUCCESS in 2m 42s\n- openstack-tox-validate http://logs.openstack.org/80/546280/3/gate/openstack-tox-validate/a9c0ede/ : SUCCESS in 2m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/546280/3/gate/build-openstack-sphinx-docs/97e169f/html/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"556e08a29f2a1380c87b7e50caa0b09d4ad73cb5","revisions":{"5120f848b9449c2f68fb32a7596ccc21481e2beb":{"kind":"REWORK","_number":1,"created":"2018-02-20 19:20:45.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/80/546280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/546280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/546280/1"}}},"commit":{"parents":[{"commit":"46b1892bc2674d1923b45dcb2dbbf07d1c68185e","subject":"Retry pypi name checking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46b1892bc2674d1923b45dcb2dbbf07d1c68185e"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-02-20 18:34:01.000000000","tz":-300},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-02-20 18:35:20.000000000","tz":-300},"subject":"support pypi-name in repository-settings","message":"support pypi-name in repository-settings\n\nAdd a repository-settings field for a name to override the name used\non the python package index.\n\nChange-Id: Ie4dc2226ea0765919a9e6eb400983260094784a7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5120f848b9449c2f68fb32a7596ccc21481e2beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5120f848b9449c2f68fb32a7596ccc21481e2beb"}]},"branch":"refs/heads/master"},"e0c03de3db6d7ce2f76ec0f40b61d3e2e258df69":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-02-20 22:56:23.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/80/546280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/546280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/546280/2"}}},"commit":{"parents":[{"commit":"805f1829a6e98338e7ca7b40e6ccdaa1dd1f2b71","subject":"Retry pypi name checking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/805f1829a6e98338e7ca7b40e6ccdaa1dd1f2b71"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-02-20 18:34:01.000000000","tz":-300},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-02-20 22:52:44.000000000","tz":-300},"subject":"support pypi-name in repository-settings","message":"support pypi-name in repository-settings\n\nAdd a repository-settings field for a name to override the name used\non the python package index.\n\nChange-Id: Ie4dc2226ea0765919a9e6eb400983260094784a7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0c03de3db6d7ce2f76ec0f40b61d3e2e258df69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0c03de3db6d7ce2f76ec0f40b61d3e2e258df69"}]},"branch":"refs/heads/master"},"556e08a29f2a1380c87b7e50caa0b09d4ad73cb5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-02-21 21:24:18.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/80/546280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/546280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/546280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/546280/3"}}},"commit":{"parents":[{"commit":"80a0c30a369022adbda9e992c10bf6adc376fa20","subject":"Retry pypi name checking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/80a0c30a369022adbda9e992c10bf6adc376fa20"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-02-20 18:34:01.000000000","tz":-300},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-02-21 21:22:49.000000000","tz":-300},"subject":"support pypi-name in repository-settings","message":"support pypi-name in repository-settings\n\nAdd a repository-settings field for a name to override the name used\non the python package index.\n\nChange-Id: Ie4dc2226ea0765919a9e6eb400983260094784a7\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/556e08a29f2a1380c87b7e50caa0b09d4ad73cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/556e08a29f2a1380c87b7e50caa0b09d4ad73cb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
