)]}'
{"id":"openstack%2Freleases~557480","triplet_id":"openstack%2Freleases~master~I478c37eacb701e97390f376226db203b4bf3d1c6","project":"openstack/releases","branch":"master","topic":"glance-policy-refactor-branch","hashtags":[],"change_id":"I478c37eacb701e97390f376226db203b4bf3d1c6","subject":"Add feature/policy-refactor branch to glance repo","status":"MERGED","created":"2018-03-28 18:54:35.000000000","updated":"2018-03-28 21:44:13.000000000","submitted":"2018-03-28 21:44:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"557480-1522273453106-6691abb4","meta_rev_id":"012ab5fae709e4067195c7950f5692136621f88b","_number":557480,"virtual_id_number":557480,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-03-28 21:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":2,"date":"2018-03-28 21:34:48.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":1,"date":"2018-03-28 21:34:48.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-28 21:34:48.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2018-03-28 21:44:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ad5d9fbf645d56e7b9a9484bb54962575a5e98e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2018-03-28 18:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f87dbee6b0bf730c9e039083fe3cb1fb2c843e81","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2018-03-28 18:55:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cb78c4ea419dec464f3640c42786c5dddd40ef8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 19:05:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-validate http://logs.openstack.org/80/557480/2/check/openstack-tox-validate/fffe100/ : FAILURE in 3m 07s\n- releases-tox-list-changes http://logs.openstack.org/80/557480/2/check/releases-tox-list-changes/f8955e1/ : SUCCESS in 3m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/557480/2/check/build-openstack-sphinx-docs/ba2887f/html/ : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3fb07782efcdedddf438c21e9cab23da908792e5","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-03-28 19:16:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5eed4a0df2bbb14f550bf502a1ce0ee04d94e972","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2018-03-28 19:59:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e44763af83ce2569c9cd09c7173c5175ae77705","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 20:06:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-validate http://logs.openstack.org/80/557480/3/check/openstack-tox-validate/d4b8775/ : SUCCESS in 3m 57s\n- releases-tox-list-changes http://logs.openstack.org/80/557480/3/check/releases-tox-list-changes/e7015cb/ : SUCCESS in 3m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/557480/3/check/build-openstack-sphinx-docs/120a705/html/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2d6e6870a43d177f76c30bafec2f9ece7acd1db8","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-03-28 21:34:48.000000000","message":"Patch Set 3: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d5a3fbe359fe96639a283e16b85587215897484f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 21:35:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a58a2ba120f33cb438e82151355de1858674b803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 21:44:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-validate http://logs.openstack.org/80/557480/3/gate/openstack-tox-validate/cc4a2c3/ : SUCCESS in 4m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/80/557480/3/gate/build-openstack-sphinx-docs/b561f57/html/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"454c4d32d3ea60bfa859af33395edc4ee5a6e050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 21:44:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9c4e291ad2041d4963734a5b824f4518ac876b34","revisions":{"35ce97fd5240c0dcf727a3332fc5a6e5bc9ed2c1":{"kind":"REWORK","_number":1,"created":"2018-03-28 18:54:35.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/557480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/557480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/557480/1"}}},"commit":{"parents":[{"commit":"7a6c61453ea3ea4c698fc68208f9f644e97827af","subject":"Merge \"Remove branch heat-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a6c61453ea3ea4c698fc68208f9f644e97827af"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2018-03-28 18:52:17.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2018-03-28 18:52:17.000000000","tz":60},"subject":"Add feature/policy-refactor branch to glane repo","message":"Add feature/policy-refactor branch to glane repo\n\nThe policy refactoring work will be significant effort touching many\ninternals of glance codebase. Creating feature branch so we can ensure\nthat the changes are stable and safe before committing to release them.\nThis will allow us to conduct the work on smaller increments that are\neasier to digest.\n\nChange-Id: I478c37eacb701e97390f376226db203b4bf3d1c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35ce97fd5240c0dcf727a3332fc5a6e5bc9ed2c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/35ce97fd5240c0dcf727a3332fc5a6e5bc9ed2c1"}]},"branch":"refs/heads/master"},"4257582b2d242cf9ca3090e23deebc29fad66379":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-03-28 18:55:10.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/557480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/557480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/557480/2"}}},"commit":{"parents":[{"commit":"7a6c61453ea3ea4c698fc68208f9f644e97827af","subject":"Merge \"Remove branch heat-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a6c61453ea3ea4c698fc68208f9f644e97827af"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2018-03-28 18:52:17.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2018-03-28 18:54:49.000000000","tz":60},"subject":"Add feature/policy-refactor branch to glance repo","message":"Add feature/policy-refactor branch to glance repo\n\nThe policy refactoring work will be significant effort touching many\ninternals of glance codebase. Creating feature branch so we can ensure\nthat the changes are stable and safe before committing to release them.\nThis will allow us to conduct the work on smaller increments that are\neasier to digest.\n\nChange-Id: I478c37eacb701e97390f376226db203b4bf3d1c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4257582b2d242cf9ca3090e23deebc29fad66379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4257582b2d242cf9ca3090e23deebc29fad66379"}]},"branch":"refs/heads/master"},"9c4e291ad2041d4963734a5b824f4518ac876b34":{"kind":"REWORK","_number":3,"created":"2018-03-28 19:59:59.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/80/557480/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/557480/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/557480/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/557480/3"}}},"commit":{"parents":[{"commit":"7a6c61453ea3ea4c698fc68208f9f644e97827af","subject":"Merge \"Remove branch heat-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a6c61453ea3ea4c698fc68208f9f644e97827af"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2018-03-28 18:52:17.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2018-03-28 19:59:27.000000000","tz":60},"subject":"Add feature/policy-refactor branch to glance repo","message":"Add feature/policy-refactor branch to glance repo\n\nThe policy refactoring work will be significant effort touching many\ninternals of glance codebase. Creating feature branch so we can ensure\nthat the changes are stable and safe before committing to release them.\nThis will allow us to conduct the work on smaller increments that are\neasier to digest.\n\nChange-Id: I478c37eacb701e97390f376226db203b4bf3d1c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c4e291ad2041d4963734a5b824f4518ac876b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c4e291ad2041d4963734a5b824f4518ac876b34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
