)]}'
{"id":"openstack%2Freleases~590476","triplet_id":"openstack%2Freleases~master~I6cf0ff8e7771367bce17061c9b21b7740516d31d","project":"openstack/releases","branch":"master","topic":"check-py-build","hashtags":[],"change_id":"I6cf0ff8e7771367bce17061c9b21b7740516d31d","subject":"add validation for sdist building","status":"MERGED","created":"2018-08-09 18:35:17.000000000","updated":"2018-08-10 09:31:34.000000000","submitted":"2018-08-10 09:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"590476-1533893494819-b9d9e23f","meta_rev_id":"dc895562e8b0337edcf34055f514e95f287e531f","_number":590476,"virtual_id_number":590476,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-08-10 09:31:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2018-08-09 19:16:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2018-08-10 09:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2018-08-10 09:19:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-09 19:16:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2018-08-10 09:19:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2018-08-10 09:31:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2eaeb965f9683e3171121baadef9541f47814f1c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-08-09 18:35:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9749b5396b51bd053b59c5c466b2788c78f22bb6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-08-09 18:37:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af15b89dee12b7c175de46872414c9978cd5e14b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-09 19:08:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/76/590476/1/check/openstack-tox-py35/a30c567/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/76/590476/1/check/openstack-tox-pep8/ecef90a/ : SUCCESS in 3m 59s\n- openstack-tox-validate http://logs.openstack.org/76/590476/1/check/openstack-tox-validate/ea6afeb/ : SUCCESS in 4m 15s\n- releases-tox-list-changes http://logs.openstack.org/76/590476/1/check/releases-tox-list-changes/6946c81/ : SUCCESS in 4m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/76/590476/1/check/build-openstack-sphinx-docs/bda485a/html/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"120262e2bb045917c96e47ae29d1089569a53418","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-08-09 19:15:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c587415a2dde84067efcb9178eb580515300b1a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-08-09 19:16:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8e19000c082aba08912c98fa3584b326e5f9e46","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2018-08-09 19:16:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4998f487d2c00628ab6e4811d218069c6f66aebd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-09 19:29:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/76/590476/2/check/openstack-tox-py35/8dcede1/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/76/590476/2/check/openstack-tox-pep8/9c44bfa/ : SUCCESS in 5m 09s\n- openstack-tox-validate http://logs.openstack.org/76/590476/2/check/openstack-tox-validate/2023781/ : SUCCESS in 5m 35s\n- releases-tox-list-changes http://logs.openstack.org/76/590476/2/check/releases-tox-list-changes/7e1ad39/ : SUCCESS in 4m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/76/590476/2/check/build-openstack-sphinx-docs/00bb6b4/html/ : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":2},{"id":"7750d206c12e0474262a0733a7527430b6ddc6f3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2018-08-10 09:19:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0a1c3e9b77e4353456d240d624e81eb67f2c58f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-10 09:19:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"722da0581873e2bf9669e44e7b4e779745799e7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-10 09:31:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7285a0e03e1820189925d03179ec0333fe7815cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-10 09:31:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/76/590476/2/gate/openstack-tox-py35/11ec61a/ : SUCCESS in 4m 30s\n- openstack-tox-validate http://logs.openstack.org/76/590476/2/gate/openstack-tox-validate/4cdd7ab/ : SUCCESS in 3m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/76/590476/2/gate/build-openstack-sphinx-docs/0d3dd03/html/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81047e0096465ee31c5c76f76dd04f3947696c06","revisions":{"4f11e8b1d503d20b5c1ae45c49998e8ca7ee6792":{"kind":"REWORK","_number":1,"created":"2018-08-09 18:35:17.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/590476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/590476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/590476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/590476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/590476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/590476/1"}}},"commit":{"parents":[{"commit":"aa6ddd51e7ec596f150477942934d3d526c29a90","subject":"Merge \"Release Heat Rocky (11.0.0.0rc1)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa6ddd51e7ec596f150477942934d3d526c29a90"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-09 18:29:04.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-09 18:34:55.000000000","tz":-240},"subject":"add validation for sdist building","message":"add validation for sdist building\n\nTry to build the sdist and wheel for python packages.\n\nChange-Id: I6cf0ff8e7771367bce17061c9b21b7740516d31d\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f11e8b1d503d20b5c1ae45c49998e8ca7ee6792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f11e8b1d503d20b5c1ae45c49998e8ca7ee6792"}]},"branch":"refs/heads/master"},"81047e0096465ee31c5c76f76dd04f3947696c06":{"kind":"REWORK","_number":2,"created":"2018-08-09 19:15:37.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/590476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/590476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/590476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/590476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/590476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/590476/2"}}},"commit":{"parents":[{"commit":"aa6ddd51e7ec596f150477942934d3d526c29a90","subject":"Merge \"Release Heat Rocky (11.0.0.0rc1)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa6ddd51e7ec596f150477942934d3d526c29a90"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-09 18:29:04.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-09 19:15:31.000000000","tz":-240},"subject":"add validation for sdist building","message":"add validation for sdist building\n\nTry to build the sdist and wheel for python packages.\n\nChange-Id: I6cf0ff8e7771367bce17061c9b21b7740516d31d\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81047e0096465ee31c5c76f76dd04f3947696c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81047e0096465ee31c5c76f76dd04f3947696c06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
