)]}'
{"id":"openstack%2Freleases~648289","triplet_id":"openstack%2Freleases~master~Ied1f6d06a33baf3d3ef3d789bbe5c7f70d967d62","project":"openstack/releases","branch":"master","topic":"feature/new_tools","hashtags":[],"change_id":"Ied1f6d06a33baf3d3ef3d789bbe5c7f70d967d62","subject":"Add wrapper script that can be used to update deliverables","status":"MERGED","created":"2019-03-27 22:46:40.000000000","updated":"2020-01-09 18:41:47.000000000","submitted":"2020-01-09 18:41:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"648289-1578595307097-69f755c7","meta_rev_id":"8f4c1df8a6af1192d8f27ce6131ffbcc4847ff7b","_number":648289,"virtual_id_number":648289,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2019-11-27 12:11:54.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 18:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-09 17:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-11-18 09:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-09 17:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 17:43:44.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-11-15 08:51:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-18 09:50:07.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-27 12:11:54.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-09 17:57:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-09 18:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc5c8dcf63daaa20bed51a36fd56e3f311da30fa","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-03-27 22:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d420b2566d25e866530cc4d83b149c78a3a04a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 22:58:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/89/648289/1/check/openstack-tox-py35/4118caf/ : SUCCESS in 8m 30s\n- openstack-tox-py36 http://logs.openstack.org/89/648289/1/check/openstack-tox-py36/79a064c/ : SUCCESS in 9m 58s\n- openstack-tox-docs http://logs.openstack.org/89/648289/1/check/openstack-tox-docs/a0bea55/html/ : SUCCESS in 6m 14s\n- openstack-tox-validate http://logs.openstack.org/89/648289/1/check/openstack-tox-validate/478470a/ : SUCCESS in 3m 59s\n- releases-tox-list-changes http://logs.openstack.org/89/648289/1/check/releases-tox-list-changes/fb486ff/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/89/648289/1/check/openstack-tox-pep8/d80656c/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1177f786412ad349806518639d02bc41bb6b3540","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-03-27 23:07:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d07be6ef6f661020fae9f7b4450cd29f1bb8ab1c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-12 04:21:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d68148814b4efef3ca457b096a2099c9a6c9d120","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-07 04:43:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9908fa1ecc94a4bccc14ae35532aeee0fdf1e1b2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-07 04:45:48.000000000","message":"Patch Set 2: Workflow-1\n\nThese are basically the tools I used to create: https://review.opendev.org/#/q/topic:wci-m1-releases\n\nSetting -W as I want to test them again as I \u0027polished\u0027 them since last night","accounts_in_message":[],"_revision_number":2},{"id":"b7d3d7a5c9412ccc03fe2e05cf2e5d23befd93e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 04:56:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py35 http://logs.openstack.org/89/648289/2/check/openstack-tox-py35/642987e/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/89/648289/2/check/openstack-tox-py36/ffd557d/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/89/648289/2/check/openstack-tox-docs/46cdaaf/html/ : SUCCESS in 7m 05s\n- openstack-tox-validate http://logs.openstack.org/89/648289/2/check/openstack-tox-validate/e7223a2/ : SUCCESS in 4m 38s\n- releases-tox-list-changes http://logs.openstack.org/89/648289/2/check/releases-tox-list-changes/d05f58c/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/89/648289/2/check/openstack-tox-pep8/515d98c/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d292d00570aa7d69fd06424c57a4f65665b82fd7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-07 17:43:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nAutomation is so beautiful.","accounts_in_message":[],"_revision_number":2},{"id":"012662040c6de46ac3916894c0976ddd9e40cc5f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-15 03:37:22.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"42b64b734c8c815102ed3c68b9155c5f76fabcd0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-15 04:19:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c7a3ea5e17476684b37ed29983484d3ef1f11950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 04:33:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1487becf78224900bc275f8f7f17ba8c : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b00ba80933aa44e5b2cec39cd985e41c : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aad890ec7eca4ffd960bdc1371d6b6bc : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1c9a7466eb456b8b85eaf70981642d : SUCCESS in 9m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eeddee3d5deb46fc8611dbbe3aeb69c1 : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/40b8a3b6ffbf4b52b6b46809d729e73f : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"b7ad1c9e43dd6d80c5a725f4107a0c333104b047","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-15 08:51:13.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nHello,\n\nLGTM but see my feedback inline:","accounts_in_message":[],"_revision_number":3},{"id":"f1543a98a212e7085ef4c9b9906a4e0914154e81","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-18 01:24:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cabd386d8102dfad5e368254318316f80c87d5d4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-11-18 01:34:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"777c2be526c6c7d41ad6dc108beb44a51293913d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 01:48:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692d2a55428e477e9405d051c33bdd6b : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f16279e58bc46dabe87e27eb030fd9e : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cdbf68bbe434651844216a7928eea22 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4936954c114267804ba4ce0aa5287c : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e35abface18c45359cc0fcb0ad47a91d : SUCCESS in 4m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e927325378624ddc979cbc3026d18298 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"454a5a443260c9d353e9c694db6e34b975d35493","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-18 09:50:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)\n\nLGTM, see inline comments for minor typos","accounts_in_message":[],"_revision_number":4},{"id":"4cad1ecfb207991e8497f1734e26605864f079cf","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-27 12:11:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af12ec5f990ea9b175d368527934a16559328e7e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-09 17:57:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nReview comments being addressed in https://review.opendev.org/701788","accounts_in_message":[],"_revision_number":4},{"id":"2564fecabb6ee0e0483b4902830bd0398c7b9d69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:57:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e69e6d71bab9e503ed643f8f105990cf408448a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 18:41:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdedfaca7894ece85d6ff834030c4c4 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bb88679cee840bdaff4c41f789409ac : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25de89f1ff6c45e49183c09003bd8260 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc403b88229441579236ce874b7c502b : SUCCESS in 8m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43fbdf377fa142dda7efb06843b37a23 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":4},{"id":"dadcd461037d3c42a5bcbee2bde179f00a78f364","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 18:41:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7045ff71b0db13b00adf2159a6f89b33a17768ba","revisions":{"219160dfbeed27ef12c51853804dc1d58bc5a133":{"kind":"REWORK","_number":1,"created":"2019-03-27 22:46:40.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/648289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/648289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/648289/1"}}},"commit":{"parents":[{"commit":"e8ca5a4d4dad2e9322d4ebac7bbac6b9b3f51ae6","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8ca5a4d4dad2e9322d4ebac7bbac6b9b3f51ae6"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-03-27 22:46:20.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-03-27 22:46:20.000000000","tz":660},"subject":"WiP: Add wrapper script that can be used to update deliverables","message":"WiP: Add wrapper script that can be used to update deliverables\n\nAt various stages we want to update lots of deliverables.  This acts as a tool to consolodate what we mean to supliment the docs and etherpad\n\nChange-Id: Ied1f6d06a33baf3d3ef3d789bbe5c7f70d967d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/219160dfbeed27ef12c51853804dc1d58bc5a133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/219160dfbeed27ef12c51853804dc1d58bc5a133"}]},"branch":"refs/heads/master"},"ed994003890dba127d0a2a297d35ae5c862ae5cc":{"kind":"REWORK","_number":2,"created":"2019-06-07 04:43:21.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/648289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/648289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/648289/2"}}},"commit":{"parents":[{"commit":"2eb4f5549d922d77010163ec3bd06ca50612af3b","subject":"Add --interactive to new-releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2eb4f5549d922d77010163ec3bd06ca50612af3b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-03-27 22:46:20.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-06-07 04:42:06.000000000","tz":600},"subject":"Add wrapper script that can be used to update deliverables","message":"Add wrapper script that can be used to update deliverables\n\nAt various stages we want to update lots of deliverables.  This acts as\na tool to consolidate what we mean to supplement the docs and etherpad\n\nStory: 2005707\nTask: 31033\n\nChange-Id: Ied1f6d06a33baf3d3ef3d789bbe5c7f70d967d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ed994003890dba127d0a2a297d35ae5c862ae5cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ed994003890dba127d0a2a297d35ae5c862ae5cc"}]},"branch":"refs/heads/master"},"d3ffb8d5273f18daf1a9055195ad2ab074f6c568":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-15 04:19:10.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/648289/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/648289/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/648289/3"}}},"commit":{"parents":[{"commit":"b18dfd6765fa1acc3d01373531a5675e66b48768","subject":"Add --interactive to new-releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b18dfd6765fa1acc3d01373531a5675e66b48768"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-03-27 22:46:20.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-14 22:35:44.000000000","tz":660},"subject":"Add wrapper script that can be used to update deliverables","message":"Add wrapper script that can be used to update deliverables\n\nAt various stages we want to update lots of deliverables.  This acts as\na tool to consolidate what we mean to supplement the docs and etherpad\n\nStory: 2005707\nTask: 31033\n\nChange-Id: Ied1f6d06a33baf3d3ef3d789bbe5c7f70d967d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3ffb8d5273f18daf1a9055195ad2ab074f6c568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3ffb8d5273f18daf1a9055195ad2ab074f6c568"}]},"branch":"refs/heads/master"},"7045ff71b0db13b00adf2159a6f89b33a17768ba":{"kind":"REWORK","_number":4,"created":"2019-11-18 01:34:38.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/648289/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/648289/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/648289/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/648289/4"}}},"commit":{"parents":[{"commit":"4cf2670c1fb36bfe93c4a3ea81f4aa10cfa42fd1","subject":"Add --interactive to new-releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4cf2670c1fb36bfe93c4a3ea81f4aa10cfa42fd1"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-03-27 22:46:20.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-11-18 01:24:00.000000000","tz":660},"subject":"Add wrapper script that can be used to update deliverables","message":"Add wrapper script that can be used to update deliverables\n\nAt various stages we want to update lots of deliverables.  This acts as\na tool to consolidate what we mean to supplement the docs and etherpad\n\nStory: 2005707\nTask: 31033\n\nChange-Id: Ied1f6d06a33baf3d3ef3d789bbe5c7f70d967d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7045ff71b0db13b00adf2159a6f89b33a17768ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7045ff71b0db13b00adf2159a6f89b33a17768ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
