)]}'
{"id":"openstack%2Freleases~650467","triplet_id":"openstack%2Freleases~master~Ibac7f787950bf23e9a75fce37a6b82ec45d7df89","project":"openstack/releases","branch":"master","topic":"eol-em","hashtags":[],"change_id":"Ibac7f787950bf23e9a75fce37a6b82ec45d7df89","subject":"Tag stable/ocata as extended maintenance","status":"MERGED","created":"2019-04-05 22:12:16.000000000","updated":"2019-04-15 14:04:34.000000000","submitted":"2019-04-12 23:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650467-1555111299186-06c4f5fb","meta_rev_id":"c47e7093d524baf7847a33d2342c9e5c424408a6","_number":650467,"virtual_id_number":650467,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-12 14:18:27.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2019-04-12 23:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-12 03:53:10.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-04-11 16:36:41.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-11 16:39:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-04-12 23:06:21.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-08 15:54:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-04-09 13:35:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-04-11 08:02:10.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-04-11 16:36:41.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-04-11 16:39:26.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-12 23:06:21.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-04-12 23:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b18e0ee8cc7b86efb1cf0e6f8404a9de9ed34cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-05 22:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d3b6844628f8e55cc24e67f787cc6b77a4f5d15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 22:41:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/650467/1/check/openstack-tox-docs/2532d14/html/ : SUCCESS in 7m 32s\n- openstack-tox-validate http://logs.openstack.org/67/650467/1/check/openstack-tox-validate/bce788c/ : SUCCESS in 13m 34s\n- releases-tox-list-changes http://logs.openstack.org/67/650467/1/check/releases-tox-list-changes/5261660/ : SUCCESS in 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5004e717fb519f1ba49dc07012cd404be9d3047d","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-04-06 19:27:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab1bb4931c518c068a455cb19cba2ffbc16ca26d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-08 05:20:49.000000000","message":"Patch Set 1:\n\nGiven the base of this series will ensure we pick the last tagged release as the target for the -em tag we will skip some commits.\n\nLooking at cinder as an example:\n\n [tony@thor cinder]$ git log --oneline --no-merges 10.0.8..origin/stable/ocata\n cce47a4c2 VNX Driver: delete_hba() instead of remove_hba()\n b1a0d62f3 RBD: Don\u0027t query Ceph on stats for exclusive pools\n fae276f0d Fix for auth version change in Brcd HTTP\n 9773f963f [Unity] Add support of removing empty host\n 034feb6d7 ScaleIO Driver - adding cache and refactoring tests\n e704e8340 Unity: fail to detach lun when auto zone enabled\n 9dbcb24f0 [Unity] Add `force detach` support\n 346671c9b Unity driver: check array OE version\n 39f1f020f [VNX]Add `force detach` support\n 5be96ca9c Empty option value maybe cause Unity driver failed to initialize\n b7490b764 VMAX Ocata - a backport caused a deadlock\n\nAll of which are after we weer supposed to make this tag.  I\u0027m kinda on the fence for ocata as we\u0027re correcting my oversight but I\u0027d like to think carefully given we\u0027re about to do just this for pike","accounts_in_message":[],"_revision_number":1},{"id":"5e3baab727b1cf5a00ac9ae51eef81aff1da2463","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-08 13:13:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"55d6a875eead40bf4272335264f7e8c4435aaddc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 13:56:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/650467/2/check/openstack-tox-docs/683c519/html/ : SUCCESS in 6m 57s\n- openstack-tox-validate http://logs.openstack.org/67/650467/2/check/openstack-tox-validate/448443e/ : SUCCESS in 19m 22s\n- releases-tox-list-changes http://logs.openstack.org/67/650467/2/check/releases-tox-list-changes/6d13c15/ : SUCCESS in 26m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c4be6a1bca1f7f846e24810ba46f17254a5f48f9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-04-08 15:41:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nManila and Nova have already the EM tag.","accounts_in_message":[],"_revision_number":2},{"id":"8ef64fa3eaa9131fd24d0114119df3abbab4ecab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-08 15:51:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ed95da1935e50602fee6c2ca0e1019b4df0f963","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-08 15:52:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da46b06c4d79dc12a35e11cc563810f22503bebd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-04-08 15:54:44.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks, too :)","accounts_in_message":[],"_revision_number":3},{"id":"5fa41063a8cefe6b2d5b20beeef085e70d063676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 16:33:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/650467/3/check/openstack-tox-docs/7fe94e5/html/ : SUCCESS in 9m 02s\n- openstack-tox-validate http://logs.openstack.org/67/650467/3/check/openstack-tox-validate/12de775/ : SUCCESS in 13m 23s\n- releases-tox-list-changes http://logs.openstack.org/67/650467/3/check/releases-tox-list-changes/96a14c4/ : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":3},{"id":"0477a0dc7a2388f26c926ecfef8521b42fe3a14b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-09 13:35:15.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks for applying it everywhere, for neutron we planned to add such tag after we made a final ocata release, but this fell off my radar.\nBetter to have it set overall!","accounts_in_message":[],"_revision_number":3},{"id":"a2c8ff3a68ddfdec101f8f8ad71e6d209b55b09d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-11 07:02:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5aff974bf01a84fa134137520b2ce595432662fe","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-11 07:03:41.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nDo you mind if I edit this myself?","accounts_in_message":[],"_revision_number":3},{"id":"0ec3e41d342ca932964d0b25a568d84eb7f14bff","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-04-11 08:02:10.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"19049a7dcc171fe8c3242c10a208c9bb34f809d8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-11 11:28:59.000000000","message":"Patch Set 3: Workflow-1\n\n\u003e (1 comment)\n\nI see. That repo doesn\u0027t even have a stable/pike branch anymore. Interesting that this wasn\u0027t tagged EOL here. That would have prevented the possibility of adding another tag like this.\n\nLooks like I will need to do a rebase to fix some merge conflicts anyway, so I will drop this one from the set.","accounts_in_message":[],"_revision_number":3},{"id":"3ac7fa23dad5e7c75b6e4ff4f670216c730a52a1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-11 11:46:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"68889bdf9b13b4a0546999039bb70723bead4f55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 12:10:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/650467/4/check/openstack-tox-docs/b2a39e1/html/ : SUCCESS in 6m 47s\n- openstack-tox-validate http://logs.openstack.org/67/650467/4/check/openstack-tox-validate/fd5ee23/ : SUCCESS in 11m 58s\n- releases-tox-list-changes http://logs.openstack.org/67/650467/4/check/releases-tox-list-changes/1ea53b7/ : SUCCESS in 23m 08s","accounts_in_message":[],"_revision_number":4},{"id":"f60005ae968f14c98b4b8bdc30789fb2cb4d0282","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-04-11 16:36:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2936520e773a276ea34a4534ce99eb80871a3bea","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-11 16:39:26.000000000","message":"Patch Set 4: Code-Review+1\n\nI didn\u0027t check everything, but it makes sense to do. Shy +1.","accounts_in_message":[],"_revision_number":4},{"id":"af20b140a457a29cf15d9b69fb7353bb59359704","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-12 03:53:10.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nLooks good to me.\n\nNot approving as I wont be able to watch for failures.","accounts_in_message":[],"_revision_number":4},{"id":"697ef71ac42ed75499de0b85d35c07a6c9878874","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-12 14:18:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd8cdd2c0276751e98f575656a4643f9bb84fc23","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-12 23:06:21.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"78dc890939fc0c3f3d36a8ed9337c845cf13f734","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 23:06:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3e710f0558b78c08e6de17f70eab673d2e4330b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 23:21:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/650467/4/gate/openstack-tox-docs/639f5e5/html/ : SUCCESS in 6m 24s\n- openstack-tox-validate http://logs.openstack.org/67/650467/4/gate/openstack-tox-validate/2d7b901/ : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5dcb6a3bfc196c2749f49566dc9adc51257b61c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 23:21:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"69d78d131e7f23415c1acd6bbd52755880ab7754","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-15 14:04:34.000000000","message":"Created a revert of this change as I2e53e7d69c1be72a5417b92b215fa972e74b4599","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d57f810ad6555528a4af039b693453b551da4cf2","revisions":{"abf2684e0a1084106375cde0cf23e912fe2656e6":{"kind":"REWORK","_number":1,"created":"2019-04-05 22:12:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/650467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/650467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/650467/1"}}},"commit":{"parents":[{"commit":"b7542ecbbedaaeb349803b59a091a6d1f95268c0","subject":"Add decorator to skip validations for EM and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7542ecbbedaaeb349803b59a091a6d1f95268c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-05 22:09:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-05 22:09:26.000000000","tz":-300},"subject":"Tag stable/ocata as extended maintenance","message":"Tag stable/ocata as extended maintenance\n\nAccording to our extended maintenance documentation [0], we stated at\nthe start of the extended maintenance phase the \"HEAD of the\nappropriate branch will be tagged as $series-em\", however, we did not\napply these tags to stable/ocata at the time. This adds ocata-em tags to\nthe ocata deliverables.\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: Ibac7f787950bf23e9a75fce37a6b82ec45d7df89\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abf2684e0a1084106375cde0cf23e912fe2656e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abf2684e0a1084106375cde0cf23e912fe2656e6"}]},"branch":"refs/heads/master"},"42af52fded55c72113d7eaa0e096a25483a6c9d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-08 13:13:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/650467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/650467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/650467/2"}}},"commit":{"parents":[{"commit":"62f2535d11eecd3afd9855c51f6aebea2a4d9c66","subject":"Add decorator to skip validations for EM and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/62f2535d11eecd3afd9855c51f6aebea2a4d9c66"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-05 22:09:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-08 13:12:35.000000000","tz":-300},"subject":"Tag stable/ocata as extended maintenance","message":"Tag stable/ocata as extended maintenance\n\nAccording to our extended maintenance documentation [0], we stated at\nthe start of the extended maintenance phase the \"HEAD of the\nappropriate branch will be tagged as $series-em\", however, we did not\napply these tags to stable/ocata at the time. This adds ocata-em tags to\nthe ocata deliverables.\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: Ibac7f787950bf23e9a75fce37a6b82ec45d7df89\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/42af52fded55c72113d7eaa0e096a25483a6c9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/42af52fded55c72113d7eaa0e096a25483a6c9d3"}]},"branch":"refs/heads/master"},"df3669e891a457524c3638dc6c2022a234096aa5":{"kind":"REWORK","_number":3,"created":"2019-04-08 15:52:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/650467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/650467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/650467/3"}}},"commit":{"parents":[{"commit":"b7542ecbbedaaeb349803b59a091a6d1f95268c0","subject":"Add decorator to skip validations for EM and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7542ecbbedaaeb349803b59a091a6d1f95268c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-05 22:09:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-08 15:52:01.000000000","tz":-300},"subject":"Tag stable/ocata as extended maintenance","message":"Tag stable/ocata as extended maintenance\n\nAccording to our extended maintenance documentation [0], we stated at\nthe start of the extended maintenance phase the \"HEAD of the\nappropriate branch will be tagged as $series-em\", however, we did not\napply these tags to stable/ocata at the time. This adds ocata-em tags to\nthe ocata deliverables.\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: Ibac7f787950bf23e9a75fce37a6b82ec45d7df89\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/df3669e891a457524c3638dc6c2022a234096aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/df3669e891a457524c3638dc6c2022a234096aa5"}]},"branch":"refs/heads/master"},"d57f810ad6555528a4af039b693453b551da4cf2":{"kind":"REWORK","_number":4,"created":"2019-04-11 11:46:17.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/650467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/650467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/650467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/650467/4"}}},"commit":{"parents":[{"commit":"898539d8fe43418aeb252b5675bab87da5101898","subject":"Add decorator to skip validations for EM and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/898539d8fe43418aeb252b5675bab87da5101898"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-05 22:09:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-11 11:46:06.000000000","tz":-300},"subject":"Tag stable/ocata as extended maintenance","message":"Tag stable/ocata as extended maintenance\n\nAccording to our extended maintenance documentation [0], we stated at\nthe start of the extended maintenance phase the \"HEAD of the\nappropriate branch will be tagged as $series-em\", however, we did not\napply these tags to stable/ocata at the time. This adds ocata-em tags to\nthe ocata deliverables.\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: Ibac7f787950bf23e9a75fce37a6b82ec45d7df89\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d57f810ad6555528a4af039b693453b551da4cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d57f810ad6555528a4af039b693453b551da4cf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
