)]}'
{"id":"openstack%2Freleases~651467","triplet_id":"openstack%2Freleases~master~I67b9f53ac4cc0e4318fb7d62ac3d389c8cafbab2","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I67b9f53ac4cc0e4318fb7d62ac3d389c8cafbab2","subject":"heat-translator 1.3.1","status":"MERGED","created":"2019-04-10 09:01:15.000000000","updated":"2019-05-10 16:11:39.000000000","submitted":"2019-05-10 16:11:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651467-1557504699955-ca9dbaec","meta_rev_id":"518e01e9fbde23bf08cfd9d6074463a4c3760c5a","_number":651467,"virtual_id_number":651467,"owner":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-05-10 16:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-06 14:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-08 15:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},{"value":1,"date":"2019-05-04 02:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":1,"date":"2019-05-04 11:30:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 15:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-08 05:10:06.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-05-08 05:09:44.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},{"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-10 09:02:54.000000000","updated_by":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"reviewer":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"state":"REVIEWER"},{"updated":"2019-04-11 09:02:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-04 02:36:57.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2019-05-04 11:30:26.000000000","updated_by":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"reviewer":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"state":"REVIEWER"},{"updated":"2019-05-08 05:09:44.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-08 05:10:06.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-08 15:20:45.000000000","updated_by":{"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},"reviewer":{"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},"state":"REVIEWER"},{"updated":"2019-05-10 15:56:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 16:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b13285cb7f7d7bb36b0e788de4bcd5d37d68403","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-04-10 09:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"543782aa978564d5aff481c7b863ee5d658e3812","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 09:14:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/651467/1/check/openstack-tox-docs/e1ee21f/html/ : SUCCESS in 6m 58s\n- openstack-tox-validate http://logs.openstack.org/67/651467/1/check/openstack-tox-validate/f2a4f26/ : FAILURE in 4m 58s\n- releases-tox-list-changes http://logs.openstack.org/67/651467/1/check/releases-tox-list-changes/b6bc4d1/ : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8155ab85484bf4fd58afed664ebf9eb9cd30d210","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-10 18:39:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"429cd95d344a727c60184646cb61ace5283426a3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-11 09:02:44.000000000","message":"Patch Set 1:\n\nYes this is a master commit -- this fix needs to be backported to stable/stein first","accounts_in_message":[],"_revision_number":1},{"id":"ff4202afcd45d782c2c57cf3414537d7cecbb122","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-16 18:18:55.000000000","message":"Patch Set 1:\n\nIs this still needed Dharmendra?","accounts_in_message":[],"_revision_number":1},{"id":"31a48e18b727881bd158e31c365732f845e53d95","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-04-29 09:10:54.000000000","message":"Patch Set 1:\n\n\u003e Is this still needed Dharmendra?\n\nYes It is needed.\nI pushed cherry pick (i.e. [1]) for stable/stein, and waiting to merge that.\n[1]: https://review.opendev.org/#/c/651635/","accounts_in_message":[],"_revision_number":1},{"id":"4a3d9c4e41eae97d3570acc9f5dd7239435772df","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-04-30 07:03:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3afe8979c149e616add719a8a4a68ac30e500af5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 07:15:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/651467/2/check/openstack-tox-docs/093e515/html/ : SUCCESS in 6m 51s\n- openstack-tox-validate http://logs.openstack.org/67/651467/2/check/openstack-tox-validate/2020041/ : SUCCESS in 3m 52s\n- releases-tox-list-changes http://logs.openstack.org/67/651467/2/check/releases-tox-list-changes/cc3a852/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d88a300142073a6c50ea93681e389116f9aae2ab","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2019-05-04 02:36:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"53b173db1b45e5a4c6955ecc40a6f1c82e04aad0","author":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"date":"2019-05-04 11:30:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d325e30ee396449bf7a4074156f20456665e636d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-06 14:25:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57663f6ee329eea5e07dfe71e08707f17ce6a138","author":{"_account_id":28349,"name":"chaiguanglong1977","email":"249386996@qq.com","username":"chaigl"},"date":"2019-05-08 15:20:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"12ee17d605466b78da8583f94d60e45522374364","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 15:56:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c685f462885c0789fe3db3957f54f109eec767c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:56:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ea70c630d9af6e11c865ba4b299ebb055ceebb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:11:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/651467/2/gate/openstack-tox-docs/75e92f8/html/ : SUCCESS in 8m 34s\n- openstack-tox-validate http://logs.openstack.org/67/651467/2/gate/openstack-tox-validate/060de65/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"518e01e9fbde23bf08cfd9d6074463a4c3760c5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:11:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e9644154553ea08269a9fd3cf36e403fb836db0c","revisions":{"4a70507ce90c67726974e1a67301386a86aba5d8":{"kind":"REWORK","_number":1,"created":"2019-04-10 09:01:15.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/67/651467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/651467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/651467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/651467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/651467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/651467/1"}}},"commit":{"parents":[{"commit":"c57395f0eee1db9741b0c9c9841080aa6f7cebe5","subject":"Merge \"Use recommended twine check for README validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c57395f0eee1db9741b0c9c9841080aa6f7cebe5"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-04-10 08:58:28.000000000","tz":0},"committer":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-04-10 09:00:45.000000000","tz":0},"subject":"heat-translator 1.3.1","message":"heat-translator 1.3.1\n\nBug Fix\n\nChange-Id: I67b9f53ac4cc0e4318fb7d62ac3d389c8cafbab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a70507ce90c67726974e1a67301386a86aba5d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4a70507ce90c67726974e1a67301386a86aba5d8"}]},"branch":"refs/heads/master"},"e9644154553ea08269a9fd3cf36e403fb836db0c":{"kind":"REWORK","_number":2,"created":"2019-04-30 07:03:36.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/67/651467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/651467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/651467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/651467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/651467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/651467/2"}}},"commit":{"parents":[{"commit":"c57395f0eee1db9741b0c9c9841080aa6f7cebe5","subject":"Merge \"Use recommended twine check for README validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c57395f0eee1db9741b0c9c9841080aa6f7cebe5"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-04-10 08:58:28.000000000","tz":0},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@india.nec.com","date":"2019-04-30 07:02:19.000000000","tz":0},"subject":"heat-translator 1.3.1","message":"heat-translator 1.3.1\n\nBug Fix\n\nChange-Id: I67b9f53ac4cc0e4318fb7d62ac3d389c8cafbab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9644154553ea08269a9fd3cf36e403fb836db0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9644154553ea08269a9fd3cf36e403fb836db0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
