)]}'
{"id":"openstack%2Freleases~652835","triplet_id":"openstack%2Freleases~master~I727699d870c2c6544ebe9f463da8753eb39f46f1","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I727699d870c2c6544ebe9f463da8753eb39f46f1","subject":"[OpenStackAnsible] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:53:54.000000000","updated":"2019-05-11 20:17:52.000000000","submitted":"2019-05-10 21:31:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652835-1557523862289-5c776a74","meta_rev_id":"7389e9343e58183985adf63892d5bce0a771601d","_number":652835,"virtual_id_number":652835,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-10 21:31:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-10 21:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 21:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 05:53:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-28 17:10:06.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-10 21:18:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 21:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7541b54163c88969b542f43338be31f0689b7df","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:53:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f15d2699601d3ffa19c3c1a49953bf496da0dc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:12:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/652835/1/check/openstack-tox-docs/1803753/html/ : SUCCESS in 7m 50s\n- openstack-tox-validate http://logs.openstack.org/35/652835/1/check/openstack-tox-validate/4fa5aee/ : FAILURE in 10m 01s\n- releases-tox-list-changes http://logs.openstack.org/35/652835/1/check/releases-tox-list-changes/5c7d924/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"11e855df9edab7ee7dfa1aef8019acf604e7883a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-24 11:49:09.000000000","message":"Patch Set 1:\n\nWhere does the  validate_series_em: release pike-em is missing openstack/openstack-ansible-security come from? I thought we remove security everywhere :/","accounts_in_message":[],"_revision_number":1},{"id":"c44e5ca0184460c4f5effc88f2d37910fe2f111a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-28 17:09:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ea3a1acf37b4b30f64d9167447643da70c7de638","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-28 17:10:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"abc169915c6b7f479fd3f2960eeabd662045159b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 17:26:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/652835/2/check/openstack-tox-docs/3cd8ede/html/ : SUCCESS in 6m 56s\n- openstack-tox-validate http://logs.openstack.org/35/652835/2/check/openstack-tox-validate/5cd481c/ : FAILURE in 5m 14s\n- releases-tox-list-changes http://logs.openstack.org/35/652835/2/check/releases-tox-list-changes/0e33f21/ : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ba318a5f1b30778daa57c2b0acab610ed6b2dc60","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-28 17:57:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d17b1b0a89b410e26fa035bbf09806583f3f7eab","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-28 21:27:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"101db44b3a658f204d92dffe78a3cd4e2558ca70","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-28 21:27:36.000000000","message":"Removed Code-Review+2 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"2ff99b393c5f6ecd56d8f8f6f0ce7d5dd917e8ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 20:57:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"588ca2777c16ef8968e992fb537eec392d2453fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:14:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/652835/3/check/openstack-tox-docs/d06f73e/html/ : SUCCESS in 7m 15s\n- openstack-tox-validate http://logs.openstack.org/35/652835/3/check/openstack-tox-validate/14463c1/ : SUCCESS in 8m 16s\n- releases-tox-list-changes http://logs.openstack.org/35/652835/3/check/releases-tox-list-changes/1ad39ae/ : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":3},{"id":"02308b60d685351006b99dec583c9fc466ce9247","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 21:18:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"83752e3934b6cee0830ba3e2956aa72df3d406bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:18:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ca73496b7b2d5c42a11b3d976c20531ea770aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:31:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b969e547a144385cacc82e2a0983eafbb20920e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 21:31:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/652835/3/gate/openstack-tox-docs/89f1ec3/html/ : SUCCESS in 6m 51s\n- openstack-tox-validate http://logs.openstack.org/35/652835/3/gate/openstack-tox-validate/761f69a/ : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6552d5f9603fecc4167e4318df391f1e5ad08a9b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-11 20:17:52.000000000","message":"Patch Set 3:\n\noh you did it. Thanks!","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a37aa9f9621b2ef6da620ac21a05c7ff9cc91d69","revisions":{"277c2b5c8f073598ccc3410b6dd7c825dc3532bc":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:53:54.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/652835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/652835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/652835/1"}}},"commit":{"parents":[{"commit":"7664e80f014c465639a3edf8a1c9b833e83d4ef7","subject":"[OpenStackAnsible] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7664e80f014c465639a3edf8a1c9b833e83d4ef7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"subject":"[OpenStackAnsible] create pike-em tag against final releases","message":"[OpenStackAnsible] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I727699d870c2c6544ebe9f463da8753eb39f46f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/277c2b5c8f073598ccc3410b6dd7c825dc3532bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/277c2b5c8f073598ccc3410b6dd7c825dc3532bc"}]},"branch":"refs/heads/master"},"07638562b8e628275cfb1f09019530982084e683":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-28 17:09:29.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/35/652835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/652835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/652835/2"}}},"commit":{"parents":[{"commit":"4936cae262e348c33f8844319421233acc2158a0","subject":"Merge \"[ironic] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4936cae262e348c33f8844319421233acc2158a0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-28 17:09:29.000000000","tz":0},"subject":"[OpenStackAnsible] create pike-em tag against final releases","message":"[OpenStackAnsible] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I727699d870c2c6544ebe9f463da8753eb39f46f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07638562b8e628275cfb1f09019530982084e683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/07638562b8e628275cfb1f09019530982084e683"}]},"branch":"refs/heads/master"},"a37aa9f9621b2ef6da620ac21a05c7ff9cc91d69":{"kind":"REWORK","_number":3,"created":"2019-05-10 20:57:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/35/652835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/652835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/652835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/652835/3"}}},"commit":{"parents":[{"commit":"4936cae262e348c33f8844319421233acc2158a0","subject":"Merge \"[ironic] create pike-em tag against final releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4936cae262e348c33f8844319421233acc2158a0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 20:51:33.000000000","tz":-300},"subject":"[OpenStackAnsible] create pike-em tag against final releases","message":"[OpenStackAnsible] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended\nmaintenance by creating the pike-em tag.\n\nIt\u0027s possible that the release isn\u0027t needed (for example if\nthere are no code changes sine the previous release, sadly\nthat\u0027s non-trivial to detect en masse).\n\nIf you are the PTL or release liaison for this team please +1\nthis review if it is ready or -1 and take over the review to\ncorrect mistakes.\n\nChange-Id: I727699d870c2c6544ebe9f463da8753eb39f46f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a37aa9f9621b2ef6da620ac21a05c7ff9cc91d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a37aa9f9621b2ef6da620ac21a05c7ff9cc91d69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
