)]}'
{"id":"openstack%2Freleases~652837","triplet_id":"openstack%2Freleases~master~Ibba5629850762d515db87c828342a9b67e5ea08c","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"Ibba5629850762d515db87c828342a9b67e5ea08c","subject":"[heat] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:54:05.000000000","updated":"2019-05-21 17:06:23.000000000","submitted":"2019-05-21 17:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652837-1558458383132-3b81190c","meta_rev_id":"d68006f285f8294390a16df9b1a88332b7ba4cd5","_number":652837,"virtual_id_number":652837,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2019-05-21 17:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-21 16:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-04-22 07:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-21 16:57:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 07:53:34.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-05-21 16:57:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-21 17:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7be2b94930078ae09c53ecc5192c359493962a5","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:54:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0f1b23d957776ee1e0f150ca31e67e160c08c2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:09:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/37/652837/1/check/openstack-tox-docs/809c678/html/ : SUCCESS in 7m 41s\n- openstack-tox-validate http://logs.openstack.org/37/652837/1/check/openstack-tox-validate/2b51efb/ : FAILURE in 5m 42s\n- releases-tox-list-changes http://logs.openstack.org/37/652837/1/check/releases-tox-list-changes/13db2ff/ : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4871167689001ae2f644d29e9a06f6d7ee9fcaac","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-04-22 07:53:34.000000000","message":"Patch Set 1: Code-Review+1\n\nseems some gate need to fix for this","accounts_in_message":[],"_revision_number":1},{"id":"b0502aad6103e84f540180d77c05ec738afd596b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 20:34:59.000000000","message":"Patch Set 1: Workflow-1\n\nNeed to hold until we decide what to do with the release job failures of the final release due to package name change.","accounts_in_message":[],"_revision_number":1},{"id":"fc871bd69c9692f4890d7a7bdcda4b00c1f58e40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-17 15:52:27.000000000","message":"Removed Workflow-1 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"44203a7e5904507be6776bb4e9283f30c7d4060a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-17 15:52:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dcd2e90ae74aaa4db7bdc2e8435a8de817e3bbcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 16:04:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/37/652837/1/check/openstack-tox-docs/a4a0b45/html/ : SUCCESS in 7m 05s\n- openstack-tox-validate http://logs.openstack.org/37/652837/1/check/openstack-tox-validate/f50e706/ : SUCCESS in 3m 54s\n- releases-tox-list-changes http://logs.openstack.org/37/652837/1/check/releases-tox-list-changes/32b88eb/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"48c6042dda84015f2d4619fc613e9e8070deb196","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-21 16:57:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae07bd7ed31ad54044da7d5997f1964e58cf63c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:57:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee03dc376b1a15f611080221ce4987028a383e47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:06:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/37/652837/1/gate/openstack-tox-docs/e8364fb/html/ : SUCCESS in 6m 54s\n- openstack-tox-validate http://logs.openstack.org/37/652837/1/gate/openstack-tox-validate/354f7e0/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"22619cdac077229bc1aa3605c34ff2aa32addc94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:06:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88effc757794d29dacad28541387e95b828cad7c","revisions":{"88effc757794d29dacad28541387e95b828cad7c":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:54:05.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/37/652837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/652837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/652837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/652837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/652837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/652837/1"}}},"commit":{"parents":[{"commit":"fbc09d5b2698b0ce26d4446e813716b0b64f6adc","subject":"[heat] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fbc09d5b2698b0ce26d4446e813716b0b64f6adc"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"subject":"[heat] create pike-em tag against final releases","message":"[heat] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ibba5629850762d515db87c828342a9b67e5ea08c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88effc757794d29dacad28541387e95b828cad7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88effc757794d29dacad28541387e95b828cad7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
