)]}'
{"id":"openstack%2Freleases~652841","triplet_id":"openstack%2Freleases~master~I37567324cdc5180a3999fc751fa0b9a462c96a22","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I37567324cdc5180a3999fc751fa0b9a462c96a22","subject":"[freezer] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:54:27.000000000","updated":"2019-05-10 18:40:08.000000000","submitted":"2019-05-10 18:40:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652841-1557513608854-121c7fd4","meta_rev_id":"ae91d58f23831e78cf18faf436195f2bd31879be","_number":652841,"virtual_id_number":652841,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":2,"date":"2019-05-10 18:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-09 20:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 18:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-04-16 05:54:27.000000000","_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 05:54:27.000000000","updated_by":{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},"reviewer":{"_account_id":22484,"name":"gengchc2","email":"geng.changcai2@zte.com.cn","username":"gengchc2"},"state":"REVIEWER"},{"updated":"2019-05-10 18:27:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 18:40:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"daeef1fdaebbc9fd2676502c698ffc086044468d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cac0591709626682d1fe1ea27b40acba8b16be8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:08:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/41/652841/1/check/openstack-tox-docs/cdcf619/html/ : SUCCESS in 7m 53s\n- openstack-tox-validate http://logs.openstack.org/41/652841/1/check/openstack-tox-validate/1f55217/ : FAILURE in 4m 43s\n- releases-tox-list-changes http://logs.openstack.org/41/652841/1/check/releases-tox-list-changes/905f756/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"58c60cf33c54fcfbe2ab3704e3cc0b15cf182805","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 19:49:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cebd20dcc5900f0c90237b8fc8ac2d113dd7e65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 20:14:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/41/652841/1/check/openstack-tox-docs/4096772/html/ : SUCCESS in 7m 01s\n- openstack-tox-validate http://logs.openstack.org/41/652841/1/check/openstack-tox-validate/f6f6a8c/ : SUCCESS in 5m 14s\n- releases-tox-list-changes http://logs.openstack.org/41/652841/1/check/releases-tox-list-changes/ab769ae/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"15732b4c432e9e8043efe669a6a9439b2aba2980","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 20:19:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"478b0e096809c40bbb38270a0c7f7be36c8bc55d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 18:27:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05d5a332d01870af37c1996d872e24e8f7ad7e19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:27:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"537d07aa19f82f8e09d702346db46f2354f9dbf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:40:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/41/652841/1/gate/openstack-tox-docs/25c8e05/html/ : SUCCESS in 7m 04s\n- openstack-tox-validate http://logs.openstack.org/41/652841/1/gate/openstack-tox-validate/78be6ed/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2e7a07455f2419d8e1e0046c04aa2e7a0f151663","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:40:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6620ff79740b30b7e77826ba1e88258d83fc4e1b","revisions":{"6620ff79740b30b7e77826ba1e88258d83fc4e1b":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:54:27.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/41/652841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/652841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/652841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/652841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/652841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/652841/1"}}},"commit":{"parents":[{"commit":"1911fab6b9d6b5528042be34e05ff8fc77edc26f","subject":"[freezer] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1911fab6b9d6b5528042be34e05ff8fc77edc26f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:17.000000000","tz":600},"subject":"[freezer] create pike-em tag against final releases","message":"[freezer] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I37567324cdc5180a3999fc751fa0b9a462c96a22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6620ff79740b30b7e77826ba1e88258d83fc4e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6620ff79740b30b7e77826ba1e88258d83fc4e1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
