)]}'
{"id":"openstack%2Freleases~652846","triplet_id":"openstack%2Freleases~master~I9216306aef6974b9275efeebf32c31631ffa63a6","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I9216306aef6974b9275efeebf32c31631ffa63a6","subject":"[tripleo] final releases for pike","status":"MERGED","created":"2019-04-16 05:55:01.000000000","updated":"2019-05-14 14:30:03.000000000","submitted":"2019-05-14 14:30:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652846-1557844203842-0de1d12d","meta_rev_id":"f59991d61997c2feec04c4d2b5f30300b21a31da","_number":652846,"virtual_id_number":652846,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-14 14:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-14 14:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-14 14:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 19:29:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-14 14:20:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-14 14:30:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4cebab6de7264ab1e30d8c1ecc2ad78b98adb97","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:55:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c056b1cbf5c109d9909f986e810e348bd18e06ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:10:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/652846/1/check/openstack-tox-docs/7a6ca8e/html/ : SUCCESS in 7m 57s\n- openstack-tox-validate http://logs.openstack.org/46/652846/1/check/openstack-tox-validate/98867eb/ : FAILURE in 9m 16s\n- releases-tox-list-changes http://logs.openstack.org/46/652846/1/check/releases-tox-list-changes/062d503/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3c07d16493d49fd64532236d167b03cbbe90cb72","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-16 14:49:40.000000000","message":"Patch Set 1:\n\nwe\u0027ll need to manually bump the puppet-tripleo and tripleo-ui versions before we can do this\n\npuppet-tripleo: https://review.openstack.org/#/c/653011/","accounts_in_message":[],"_revision_number":1},{"id":"184666964bae496214c27236968f9c4d1dbae7c7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-16 15:44:26.000000000","message":"Patch Set 1:\n\nhttps://review.openstack.org/#/c/653029/ for tripleo-ui","accounts_in_message":[],"_revision_number":1},{"id":"a09e03d90581711bf6f178c4dab779e5994f863c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:49:58.000000000","message":"Patch Set 1:\n\nThanks Alex","accounts_in_message":[],"_revision_number":1},{"id":"f5b89e64d4d6d5f98eb6355ed971511fb58d43b1","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:50:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c6bf5cb46cf71ddc6d60266336f42b320d54cf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 07:23:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/652846/2/check/openstack-tox-docs/0f124f2/html/ : SUCCESS in 9m 37s\n- openstack-tox-validate http://logs.openstack.org/46/652846/2/check/openstack-tox-validate/b827c30/ : FAILURE in 10m 00s\n- releases-tox-list-changes http://logs.openstack.org/46/652846/2/check/releases-tox-list-changes/bc88471/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"794e25eab8cc107e8478ad934377316ac7235bd3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-18 06:58:01.000000000","message":"Patch Set 2:\n\nI\u0027ll work through the errors[1] next week:\n\n deliverables/pike/instack.yaml: validate_release_type: openstack-infra/project-config/zuul.d/projects.yaml no release job specified for openstack/instack, one of [\u0027publish-to-pypi\u0027] needs to be included in [\u0027openstack-python-jobs\u0027, \u0027check-requirements\u0027] or no release will be published\n deliverables/pike/tripleo-image-elements.yaml: validate_pypi_readme: README check for openstack/tripleo-image-elements failed: Command \u0027[\u0027twine\u0027, \u0027check\u0027, \u0027/home/zuul/src/git.openstack.org/openstack/releases/.tox/validate/tmp/releases-55m2u2ll/openstack/tripleo-image-elements/dist/*\u0027]\u0027 returned non-zero exit status 1.\n deliverables/pike/tripleo-ui.yaml: validate_version_numbers: openstack/tripleo-ui package.json contains \"7.4.9\" but is being tagged \"7.4.8\"\n deliverables/pike/tripleo-ui.yaml: validate_new_releases_at_end: new release 7.4.8 must be listed last, with one new release per patch\n deliverables/pike/tripleo-ui.yaml: validate_release_branch_membership: openstack/tripleo-ui 3db0b4f298cc6a991627f473b9ba246a64ded09e receiving 7.4.10 is not a descendant of 7.4.8\n deliverables/pike/tripleo-validations.yaml: validate_pypi_readme: README check for openstack/tripleo-validations failed: Command \u0027[\u0027twine\u0027, \u0027check\u0027, \u0027/home/zuul/src/git.openstack.org/openstack/releases/.tox/validate/tmp/releases-55m2u2ll/openstack/tripleo-validations/dist/*\u0027]\u0027 returned non-zero exit status 1.\n\n\n[1] http://logs.openstack.org/46/652846/2/check/openstack-tox-validate/b827c30/job-output.txt.gz#_2019-04-17_07_22_17_674801","accounts_in_message":[],"_revision_number":2},{"id":"e74c070e56ec68ca046016f1ac02673c8cd59cb8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-19 00:48:13.000000000","message":"Patch Set 2:\n\nAt least instack is a retired repo, so not sure if it\u0027s worth fixing the release job for that. This might be a case where it\u0027s better just to tag the last release as -em.","accounts_in_message":[],"_revision_number":2},{"id":"812c6a849db039d732a7cc8ac81a5e2e9ffaef5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 19:50:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa03cf9e9426da835bf7932c891e3e397d7c2181","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:00:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/652846/3/check/openstack-tox-docs/9834a12/html/ : SUCCESS in 6m 40s\n- openstack-tox-validate http://logs.openstack.org/46/652846/3/check/openstack-tox-validate/55e60bd/ : FAILURE in 9m 14s\n- releases-tox-list-changes http://logs.openstack.org/46/652846/3/check/releases-tox-list-changes/f4b9ac3/ : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":3},{"id":"568fd716c0ea6fbf6eaa34c2e32db32719efa019","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-13 18:06:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f589b9b483beb5e7fa89d85bc40ebfa2a67af184","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-13 18:07:28.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"184fe26abdb60794b498327e392926b16663d68f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 18:08:20.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"05d30ac3b5c85782d57e82b90f03b37a7016012c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-13 18:12:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7a57b34c416ce243be46dc338c1a581c76e4f2ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 18:25:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/46/652846/4/check/openstack-tox-docs/161d514/html/ : SUCCESS in 7m 49s\n- openstack-tox-validate http://logs.openstack.org/46/652846/4/check/openstack-tox-validate/6a2d715/ : FAILURE in 8m 07s\n- releases-tox-list-changes http://logs.openstack.org/46/652846/4/check/releases-tox-list-changes/97d9eec/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"1599d92d7c5aee4fdff44b5a66136d65708678fa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 18:29:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5a8a8a7968e283bb1278a9940aaa5bea60c002ca","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-13 19:29:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"06cdd05b85f18ccff40ccabfc303f61a22a96569","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 19:38:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36d483963e60306357609ea741e759c99065a6f2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 13:50:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"29c3d8248d691799c168fd759b41431738750469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:08:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/652846/5/check/openstack-tox-docs/0d53b5f/html/ : SUCCESS in 7m 23s\n- openstack-tox-validate http://logs.openstack.org/46/652846/5/check/openstack-tox-validate/8b3c7e5/ : SUCCESS in 10m 24s\n- releases-tox-list-changes http://logs.openstack.org/46/652846/5/check/releases-tox-list-changes/390d65c/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":5},{"id":"8534bc29dec41302a12f957383d67f26cc2e2d1d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-14 14:20:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4a26a22d20a67887e07576ec73630e1f7e1f8b4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:21:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c80a86950185ef538dbc527832d73870ddf9d035","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:30:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"bf5a37fbdaeec4f8385d3532aabfaf0735eb5211","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 14:30:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/46/652846/5/gate/openstack-tox-docs/5b3697c/html/ : SUCCESS in 6m 35s\n- openstack-tox-validate http://logs.openstack.org/46/652846/5/gate/openstack-tox-validate/ff93363/ : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"02bc275c3c87820d9de191f6dbcf2f5c421b16ba","revisions":{"66b12d8937aac7d0bc6edebd7d7ceb582a62f3a1":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:55:01.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/46/652846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/652846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/652846/1"}}},"commit":{"parents":[{"commit":"a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df","subject":"Merge \"Branch roles that were added during stein cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4c7deef9b70bd1ed8a96ef1679788e1cd58f2df"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:08.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:08.000000000","tz":600},"subject":"[tripleo] final releases for pike","message":"[tripleo] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66b12d8937aac7d0bc6edebd7d7ceb582a62f3a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66b12d8937aac7d0bc6edebd7d7ceb582a62f3a1"}]},"branch":"refs/heads/master"},"87eb854a894b0e529393422ba64519fa0b553bef":{"kind":"REWORK","_number":2,"created":"2019-04-17 06:50:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/46/652846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/652846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/652846/2"}}},"commit":{"parents":[{"commit":"78afdabc0c2dab74a0f513a7cb7c5777d211854e","subject":"Merge \"[octavia] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78afdabc0c2dab74a0f513a7cb7c5777d211854e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:08.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 06:48:16.000000000","tz":600},"subject":"[tripleo] final releases for pike","message":"[tripleo] final releases for pike\n\nThis is a procedural review to tag a final release for each deliverables ; and\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87eb854a894b0e529393422ba64519fa0b553bef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87eb854a894b0e529393422ba64519fa0b553bef"}]},"branch":"refs/heads/master"},"0879847520b7e43e3cbf3669948350f67802f48d":{"kind":"REWORK","_number":3,"created":"2019-05-10 19:50:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/46/652846/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/652846/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/652846/3"}}},"commit":{"parents":[{"commit":"78afdabc0c2dab74a0f513a7cb7c5777d211854e","subject":"Merge \"[octavia] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78afdabc0c2dab74a0f513a7cb7c5777d211854e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:08.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 19:50:24.000000000","tz":-300},"subject":"[tripleo] final releases for pike","message":"[tripleo] final releases for pike\n\nThis is a procedural review to tag a final release for each\ndeliverables.\n\nIt\u0027s possible that the release isn\u0027t needed (for example if\nthere are no code changes sine the previous release, sadly\nthat\u0027s non-trivial to detect en masse).\n\nIf you are the PTL or release liaison for this team please +1\nthis review if it is ready or -1 and take over the review to\ncorrect mistakes.\n\nChange-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0879847520b7e43e3cbf3669948350f67802f48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0879847520b7e43e3cbf3669948350f67802f48d"}]},"branch":"refs/heads/master"},"0cbba409e8d635f90f10ec854d685c2cef579025":{"kind":"REWORK","_number":4,"created":"2019-05-13 18:06:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/46/652846/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/652846/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/652846/4"}}},"commit":{"parents":[{"commit":"78afdabc0c2dab74a0f513a7cb7c5777d211854e","subject":"Merge \"[octavia] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78afdabc0c2dab74a0f513a7cb7c5777d211854e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:08.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-13 18:06:16.000000000","tz":-360},"subject":"[tripleo] final releases for pike","message":"[tripleo] final releases for pike\n\nThis is a procedural review to tag a final release for each\ndeliverables.\n\nIt\u0027s possible that the release isn\u0027t needed (for example if\nthere are no code changes sine the previous release, sadly\nthat\u0027s non-trivial to detect en masse).\n\nIf you are the PTL or release liaison for this team please +1\nthis review if it is ready or -1 and take over the review to\ncorrect mistakes.\n\nDepends-On: https://review.opendev.org/#/c/658877/\nChange-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cbba409e8d635f90f10ec854d685c2cef579025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cbba409e8d635f90f10ec854d685c2cef579025"}]},"branch":"refs/heads/master"},"02bc275c3c87820d9de191f6dbcf2f5c421b16ba":{"kind":"REWORK","_number":5,"created":"2019-05-14 13:50:33.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/46/652846/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/46/652846/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/46/652846/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/46/652846/5"}}},"commit":{"parents":[{"commit":"78afdabc0c2dab74a0f513a7cb7c5777d211854e","subject":"Merge \"[octavia] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78afdabc0c2dab74a0f513a7cb7c5777d211854e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:08.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-14 13:50:29.000000000","tz":-360},"subject":"[tripleo] final releases for pike","message":"[tripleo] final releases for pike\n\nThis is a procedural review to tag a final release for each\ndeliverables.\n\nIt\u0027s possible that the release isn\u0027t needed (for example if\nthere are no code changes sine the previous release, sadly\nthat\u0027s non-trivial to detect en masse).\n\nIf you are the PTL or release liaison for this team please +1\nthis review if it is ready or -1 and take over the review to\ncorrect mistakes.\n\nDepends-On: https://review.opendev.org/#/c/658877/\nChange-Id: I9216306aef6974b9275efeebf32c31631ffa63a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/02bc275c3c87820d9de191f6dbcf2f5c421b16ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/02bc275c3c87820d9de191f6dbcf2f5c421b16ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
