)]}'
{"id":"openstack%2Freleases~652847","triplet_id":"openstack%2Freleases~master~Ia12fbe9479a7f2556f0d1e58868643acf84748b7","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"Ia12fbe9479a7f2556f0d1e58868643acf84748b7","subject":"[tripleo] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:55:01.000000000","updated":"2019-05-16 16:56:21.000000000","submitted":"2019-05-16 16:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652847-1558025781529-4a183efd","meta_rev_id":"48068edd6187069d6774b85d75e6192299a799eb","_number":652847,"virtual_id_number":652847,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-16 16:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-16 15:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-16 16:32:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 15:09:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-16 16:32:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-16 16:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68179a0fe15a8e97e1cc758c7f9b7f792fd51932","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:55:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b556c002510c7e1330738c6d199bb0e8090d434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:08:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/1/check/openstack-tox-docs/141dc65/html/ : SUCCESS in 7m 28s\n- openstack-tox-validate http://logs.openstack.org/47/652847/1/check/openstack-tox-validate/85945f8/ : FAILURE in 8m 01s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/1/check/releases-tox-list-changes/649d035/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f8718d1300aec8bf9dbfc537f6e36a6520b3099d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:50:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d6aa09dc4135701f1f1805835823d9adee35a4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 07:19:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/2/check/openstack-tox-docs/49a47bc/ : POST_FAILURE in 6m 02s\n- openstack-tox-validate http://logs.openstack.org/47/652847/2/check/openstack-tox-validate/8a38d68/ : FAILURE in 5m 27s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/2/check/releases-tox-list-changes/5d7be25/ : FAILURE in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9a09668408f8d084b017b7f1da613756d785c7ec","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 15:03:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"59d539871a474c120c2d404d4389bb2201c16537","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-14 15:05:47.000000000","message":"Patch Set 3:\n\nThere was a post-release failure on one of these:\n\nhttp://logs.openstack.org/8a/8a26f87a18f1948a8021181fc9c28c306d6c682e/release/release-openstack-puppet/64a54ef/job-output.txt.gz#_2019-05-14_14_44_49_238988\n\nNot sure if we need to sort that out and do one more release, or if we can safely ignore that one.","accounts_in_message":[],"_revision_number":3},{"id":"98433f6e8e9c348bd04cd61bf0fa054489d000c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 15:09:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/3/check/openstack-tox-docs/3ca934b/ : POST_FAILURE in 3m 58s\n- openstack-tox-validate http://logs.openstack.org/47/652847/3/check/openstack-tox-validate/470e9f5/ : FAILURE in 3m 52s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/3/check/releases-tox-list-changes/28d1dfe/ : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"066981eeeb51da52463f456d509edf708bda7951","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 15:09:41.000000000","message":"Patch Set 3: Code-Review-1\n\nI can fixup the metadata and rerelease puppet-tripleo.  I also need to take a look at paunch and os-refresh-config as they seemed to have been dropped from the last round of pike releases","accounts_in_message":[],"_revision_number":3},{"id":"b759542b0fa6c4c330899d25ae2f6a6ffa6d17f3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 16:47:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a4a53a7f455eda64a85b20f37936776c75629678","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 16:48:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2b3effd7f64a07dbadf43711af99d042571d7735","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 16:59:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/5/check/openstack-tox-docs/94c18d7/ : POST_FAILURE in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/47/652847/5/check/openstack-tox-validate/5061497/ : FAILURE in 3m 45s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/5/check/releases-tox-list-changes/be8303f/ : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":5},{"id":"29585037a99f8a41747e70c789add1d1c8b6a692","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 21:20:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"88b4a10a6fe3a0bcb1da2ca1bd5e50ac9fc447eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 21:35:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/6/check/openstack-tox-docs/164c3c8/ : POST_FAILURE in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/47/652847/6/check/openstack-tox-validate/d2cd372/ : FAILURE in 4m 10s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/6/check/releases-tox-list-changes/108b9cf/ : FAILURE in 5m 08s","accounts_in_message":[],"_revision_number":6},{"id":"eabc14e069717a3a5f099c4b18a4388eee6d9f67","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-14 22:38:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e4506cd0382550d3466be47e9c7453c392c584cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:45:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/7/check/openstack-tox-docs/6ecca4f/html/ : SUCCESS in 6m 00s\n- openstack-tox-validate http://logs.openstack.org/47/652847/7/check/openstack-tox-validate/bf98cd9/ : FAILURE in 4m 03s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/7/check/releases-tox-list-changes/7c69017/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":7},{"id":"c9916d35c450f08022933e33e2921e7aef816e7d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-15 16:47:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1a28b20252756669e21c561399a63175cf459db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:58:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/8/check/openstack-tox-docs/04c4694/html/ : SUCCESS in 7m 00s\n- openstack-tox-validate http://logs.openstack.org/47/652847/8/check/openstack-tox-validate/187e49a/ : FAILURE in 5m 16s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/8/check/releases-tox-list-changes/80f95f8/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":8},{"id":"07e470614580618d9de574ea98333ad57f273327","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 06:44:26.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"eea58b348092b64573a61cddefa51e01251fa2c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:57:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/8/check/openstack-tox-docs/63f574f/html/ : SUCCESS in 7m 42s\n- openstack-tox-validate http://logs.openstack.org/47/652847/8/check/openstack-tox-validate/2f071fc/ : SUCCESS in 4m 49s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/8/check/releases-tox-list-changes/4e2a563/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":8},{"id":"04119986bbee7401179a48de4c538f808943eec0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-16 06:58:09.000000000","message":"Topic changed from tripleo-pike-eol to pike-em","accounts_in_message":[],"_revision_number":8},{"id":"5d5081a2a0cd2778c8d8987b56f4ca58930ac00e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 15:28:09.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)\n\nMost look good, but an issue with a few of the deliverables.","accounts_in_message":[],"_revision_number":8},{"id":"93eadda6f460ba4d4a2f6f9baa4d0bd47579a591","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-16 15:38:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ba895b8de21b7fd76efd66d0375d30e46764d151","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 15:39:29.000000000","message":"Patch Set 9: Code-Review+2\n\nNeed to wait for zuul results, but everything checks out from my view. Thanks Alex!","accounts_in_message":[],"_revision_number":9},{"id":"357e0de1847592716625c5fa9fac484859830872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 15:50:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/9/check/openstack-tox-docs/66b853e/html/ : SUCCESS in 6m 52s\n- openstack-tox-validate http://logs.openstack.org/47/652847/9/check/openstack-tox-validate/f9a99e6/ : SUCCESS in 5m 05s\n- releases-tox-list-changes http://logs.openstack.org/47/652847/9/check/releases-tox-list-changes/495237d/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":9},{"id":"778ba6378ec77acaaf4bf94e6e17874249e00177","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 15:56:32.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e2ce582580e79eb8a9eff3450e479fbf1fc30806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 15:56:45.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a913ec44d52139ede2df477b87f4b52bc1c5c2b6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 15:57:24.000000000","message":"Patch Set 9: -Workflow\n\nOh, might be a release job issue that could impact tripleo-ui, so going to wait to get confirmation from infra first.","accounts_in_message":[],"_revision_number":9},{"id":"5551295d9381e54bd762156a0412c98951da350f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:09:02.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/9/gate/openstack-tox-docs/c3c1a51/html/ : SUCCESS in 7m 09s\n- openstack-tox-validate http://logs.openstack.org/47/652847/9/gate/openstack-tox-validate/e31e500/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":9},{"id":"a5924accf0a82f92b0f36d8e345f8ba6ea136b9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 16:32:32.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"14ef93d7d9a12522feb98cd3365993be1bee4090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:32:45.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3807d4e63814003b8e602e7abac2b758b44df6ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:56:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"48068edd6187069d6774b85d75e6192299a799eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:56:21.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/47/652847/9/gate/openstack-tox-docs/7956f85/html/ : SUCCESS in 7m 28s\n- openstack-tox-validate http://logs.openstack.org/47/652847/9/gate/openstack-tox-validate/d8e88fc/ : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"cf65dddab3cc67d79255db4d9d76df3f407ec739","revisions":{"23fc47e13941f607c08fc29340f7edd53a50bcac":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:55:01.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/47/652847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/1"}}},"commit":{"parents":[{"commit":"66b12d8937aac7d0bc6edebd7d7ceb582a62f3a1","subject":"[tripleo] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66b12d8937aac7d0bc6edebd7d7ceb582a62f3a1"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23fc47e13941f607c08fc29340f7edd53a50bcac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23fc47e13941f607c08fc29340f7edd53a50bcac"}]},"branch":"refs/heads/master"},"1abf270f07fb6176bf865b10c1a0188f89abcb6d":{"kind":"REWORK","_number":2,"created":"2019-04-17 06:50:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/47/652847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/2"}}},"commit":{"parents":[{"commit":"87eb854a894b0e529393422ba64519fa0b553bef","subject":"[tripleo] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87eb854a894b0e529393422ba64519fa0b553bef"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 06:49:33.000000000","tz":600},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1abf270f07fb6176bf865b10c1a0188f89abcb6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1abf270f07fb6176bf865b10c1a0188f89abcb6d"}]},"branch":"refs/heads/master"},"978bd855191c09f2e4f66ea8c8d512fe746c31e9":{"kind":"REWORK","_number":3,"created":"2019-05-14 15:03:45.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/3"}}},"commit":{"parents":[{"commit":"4ace26fbf93cede08b9f868ed5b1d2e418cc5d31","subject":"Merge \"[tripleo] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ace26fbf93cede08b9f868ed5b1d2e418cc5d31"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-14 15:03:39.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/978bd855191c09f2e4f66ea8c8d512fe746c31e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/978bd855191c09f2e4f66ea8c8d512fe746c31e9"}]},"branch":"refs/heads/master"},"5a3c807ca29d5d4a7abf400f4895eef7db73bd65":{"kind":"REWORK","_number":4,"created":"2019-05-14 16:47:16.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/4"}}},"commit":{"parents":[{"commit":"c132381764ff4ddb7cb6d8397e16bac718b36211","subject":"Final pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c132381764ff4ddb7cb6d8397e16bac718b36211"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-14 16:47:05.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a3c807ca29d5d4a7abf400f4895eef7db73bd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a3c807ca29d5d4a7abf400f4895eef7db73bd65"}]},"branch":"refs/heads/master"},"37c28c5b498602cb2ce2abc0dc434f2b83133e31":{"kind":"REWORK","_number":5,"created":"2019-05-14 16:48:27.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/5"}}},"commit":{"parents":[{"commit":"c132381764ff4ddb7cb6d8397e16bac718b36211","subject":"Final pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c132381764ff4ddb7cb6d8397e16bac718b36211"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-14 16:48:20.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37c28c5b498602cb2ce2abc0dc434f2b83133e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37c28c5b498602cb2ce2abc0dc434f2b83133e31"}]},"branch":"refs/heads/master"},"b9bb0b47402cca0ca320ae811a1a5297078d1c82":{"kind":"REWORK","_number":6,"created":"2019-05-14 21:20:18.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/6"}}},"commit":{"parents":[{"commit":"a8e2fe718506b6abce5219b21380fd12700ab2a5","subject":"Final pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8e2fe718506b6abce5219b21380fd12700ab2a5"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-14 21:20:11.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9bb0b47402cca0ca320ae811a1a5297078d1c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b9bb0b47402cca0ca320ae811a1a5297078d1c82"}]},"branch":"refs/heads/master"},"8e182203a7b3d2395563f79b138d28999c686609":{"kind":"REWORK","_number":7,"created":"2019-05-14 22:38:19.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/7"}}},"commit":{"parents":[{"commit":"a8e2fe718506b6abce5219b21380fd12700ab2a5","subject":"Final pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8e2fe718506b6abce5219b21380fd12700ab2a5"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-14 21:44:04.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e182203a7b3d2395563f79b138d28999c686609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e182203a7b3d2395563f79b138d28999c686609"}]},"branch":"refs/heads/master"},"2da0869a71c8135993a5350af01e2a9ecc3a3201":{"kind":"REWORK","_number":8,"created":"2019-05-15 16:47:28.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/8"}}},"commit":{"parents":[{"commit":"a8e2fe718506b6abce5219b21380fd12700ab2a5","subject":"Final pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8e2fe718506b6abce5219b21380fd12700ab2a5"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-15 16:46:31.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2da0869a71c8135993a5350af01e2a9ecc3a3201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2da0869a71c8135993a5350af01e2a9ecc3a3201"}]},"branch":"refs/heads/master"},"cf65dddab3cc67d79255db4d9d76df3f407ec739":{"kind":"REWORK","_number":9,"created":"2019-05-16 15:38:06.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/47/652847/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/652847/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/652847/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/652847/9"}}},"commit":{"parents":[{"commit":"a8e2fe718506b6abce5219b21380fd12700ab2a5","subject":"Final pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8e2fe718506b6abce5219b21380fd12700ab2a5"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-16 15:38:02.000000000","tz":-360},"subject":"[tripleo] create pike-em tag against final releases","message":"[tripleo] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia12fbe9479a7f2556f0d1e58868643acf84748b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf65dddab3cc67d79255db4d9d76df3f407ec739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf65dddab3cc67d79255db4d9d76df3f407ec739"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
