)]}'
{"id":"openstack%2Freleases~652855","triplet_id":"openstack%2Freleases~master~I1f39fbe2ea916bef45e5c696727825647015e5c7","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I1f39fbe2ea916bef45e5c696727825647015e5c7","subject":"[monasca] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:55:47.000000000","updated":"2019-05-17 13:48:28.000000000","submitted":"2019-05-17 13:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652855-1558100908273-4f1731f4","meta_rev_id":"96a5f3fb0992eca1d1820e8ff88e9c1594ae665e","_number":652855,"virtual_id_number":652855,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-05-17 13:48:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-17 11:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-17 13:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-17 11:51:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-17 13:40:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 11:23:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-17 11:51:52.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-05-17 13:40:28.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-05-17 13:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acc90a3b3029f964e7d459e027f3cacc40076bd8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:55:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef6d5a6993f8fd1d4197c46f750263a66dce6efb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:08:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/652855/1/check/openstack-tox-docs/92fb338/html/ : SUCCESS in 6m 53s\n- openstack-tox-validate http://logs.openstack.org/55/652855/1/check/openstack-tox-validate/4006315/ : FAILURE in 5m 34s\n- releases-tox-list-changes http://logs.openstack.org/55/652855/1/check/releases-tox-list-changes/a256578/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"28e5eaa4ec7b06ccc87b4d56b74f3c058a21ccc2","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 05:16:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"357631bc8188c92e48111ba42471585c68d1fccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 05:26:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/652855/2/check/openstack-tox-docs/2d68f70/html/ : SUCCESS in 6m 46s\n- openstack-tox-validate http://logs.openstack.org/55/652855/2/check/openstack-tox-validate/4aa0ce5/ : FAILURE in 5m 23s\n- releases-tox-list-changes http://logs.openstack.org/55/652855/2/check/releases-tox-list-changes/9472463/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"71e2225d8cf0ba7920181a031354ad106400a5a3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-18 05:29:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2b5ae9606fd11d821e715cb050e3e1d4bda2520f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 05:44:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/652855/3/check/openstack-tox-docs/12cf891/html/ : SUCCESS in 7m 09s\n- openstack-tox-validate http://logs.openstack.org/55/652855/3/check/openstack-tox-validate/5f86565/ : FAILURE in 5m 55s\n- releases-tox-list-changes http://logs.openstack.org/55/652855/3/check/releases-tox-list-changes/b9916c7/ : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":3},{"id":"1167a9566ca48a2004622e11bb872687c4e3070c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-16 12:31:19.000000000","message":"Patch Set 3:\n\nI\u0027ve cherry-picked changes back to Pike to fix Readme errors. Should be merged soon. I\u0027ll update then the hashes.\n\nhttps://review.opendev.org/#/q/branch:stable/pike+topic:convert_readme","accounts_in_message":[],"_revision_number":3},{"id":"e3856270aaf809e182b35c27d744401e96e71364","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-16 22:47:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"db63e8583a9a591a5fa0e8597779755f1cbe3819","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:56:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/652855/4/check/openstack-tox-docs/fd7da24/ : POST_FAILURE in 4m 01s\n- openstack-tox-validate http://logs.openstack.org/55/652855/4/check/openstack-tox-validate/1bec09e/ : FAILURE in 5m 05s\n- releases-tox-list-changes http://logs.openstack.org/55/652855/4/check/releases-tox-list-changes/b04b915/ : FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":4},{"id":"8c7f2d6fa32208bf32ccc494a0c6e8ef3de0c42f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-17 00:03:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b1288ec8e6f908be9436adc1e11e952b3cf0cbcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 00:21:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/652855/5/check/openstack-tox-docs/67ec761/html/ : SUCCESS in 6m 24s\n- openstack-tox-validate http://logs.openstack.org/55/652855/5/check/openstack-tox-validate/30bdedc/ : SUCCESS in 4m 01s\n- releases-tox-list-changes http://logs.openstack.org/55/652855/5/check/releases-tox-list-changes/0f843b0/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":5},{"id":"deec82bd5240e809a72b641407d4406b2b9c4850","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-17 11:23:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fac29a847ac4441d51e45c7f21c63b5288e04980","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-17 11:51:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"81c825245db314ebf570c8c302b3f9dfcee49d33","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-05-17 13:40:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b1a578f4184343236cba6fa534daa3e2247ba67a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 13:40:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"06e94090f8f414937f0172560a35f0e7b4ff01a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 13:48:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ea0ca76c85794f207aa242fb14382808c15d4e9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 13:48:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/652855/5/gate/openstack-tox-docs/cea08ea/html/ : SUCCESS in 5m 57s\n- openstack-tox-validate http://logs.openstack.org/55/652855/5/gate/openstack-tox-validate/32d58d6/ : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0245c245604c9ed8628cec2e7697c469b1e4c4bd","revisions":{"ee1550e027b0db26af26587c4e9c723356e7c1b5":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:55:47.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/55/652855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/652855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/652855/1"}}},"commit":{"parents":[{"commit":"5caed719272f6642e2f58c4572927f7ec13e1992","subject":"[monasca] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5caed719272f6642e2f58c4572927f7ec13e1992"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[monasca] create pike-em tag against final releases","message":"[monasca] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1f39fbe2ea916bef45e5c696727825647015e5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee1550e027b0db26af26587c4e9c723356e7c1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee1550e027b0db26af26587c4e9c723356e7c1b5"}]},"branch":"refs/heads/master"},"0204c6c35dae5604c9fafec5a630061ae4d3c68b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-17 05:16:49.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/55/652855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/652855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/652855/2"}}},"commit":{"parents":[{"commit":"f91c57ec33d0a9cf946f08158284e6e6db886512","subject":"[monasca] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f91c57ec33d0a9cf946f08158284e6e6db886512"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 05:16:39.000000000","tz":600},"subject":"[monasca] create pike-em tag against final releases","message":"[monasca] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1f39fbe2ea916bef45e5c696727825647015e5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0204c6c35dae5604c9fafec5a630061ae4d3c68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0204c6c35dae5604c9fafec5a630061ae4d3c68b"}]},"branch":"refs/heads/master"},"537326c342a978c3ab5a6f04d70c42a8968a4a3d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-18 05:29:44.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/55/652855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/652855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/652855/3"}}},"commit":{"parents":[{"commit":"6db4e9fae0608c823a5b0ce180715b98b24e8316","subject":"[monasca] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6db4e9fae0608c823a5b0ce180715b98b24e8316"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-18 05:27:27.000000000","tz":600},"subject":"[monasca] create pike-em tag against final releases","message":"[monasca] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1f39fbe2ea916bef45e5c696727825647015e5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/537326c342a978c3ab5a6f04d70c42a8968a4a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/537326c342a978c3ab5a6f04d70c42a8968a4a3d"}]},"branch":"refs/heads/master"},"d2a462a142ad870f0da2f7fd5f376968af0afec2":{"kind":"REWORK","_number":4,"created":"2019-05-16 22:47:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/55/652855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/652855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/652855/4"}}},"commit":{"parents":[{"commit":"8d75596601741c4f36b130a3a9dc2652334a0ba4","subject":"Merge \"[monasca] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d75596601741c4f36b130a3a9dc2652334a0ba4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-16 22:47:23.000000000","tz":-300},"subject":"[monasca] create pike-em tag against final releases","message":"[monasca] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1f39fbe2ea916bef45e5c696727825647015e5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d2a462a142ad870f0da2f7fd5f376968af0afec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d2a462a142ad870f0da2f7fd5f376968af0afec2"}]},"branch":"refs/heads/master"},"0245c245604c9ed8628cec2e7697c469b1e4c4bd":{"kind":"REWORK","_number":5,"created":"2019-05-17 00:03:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/55/652855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/652855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/652855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/652855/5"}}},"commit":{"parents":[{"commit":"8d75596601741c4f36b130a3a9dc2652334a0ba4","subject":"Merge \"[monasca] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d75596601741c4f36b130a3a9dc2652334a0ba4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-17 00:03:15.000000000","tz":-300},"subject":"[monasca] create pike-em tag against final releases","message":"[monasca] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I1f39fbe2ea916bef45e5c696727825647015e5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0245c245604c9ed8628cec2e7697c469b1e4c4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0245c245604c9ed8628cec2e7697c469b1e4c4bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
