)]}'
{"id":"openstack%2Freleases~652865","triplet_id":"openstack%2Freleases~master~I6bc59c8b97c852111c54be155c4ddb4f27cb9c9c","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I6bc59c8b97c852111c54be155c4ddb4f27cb9c9c","subject":"[murano] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:56:45.000000000","updated":"2019-04-23 01:02:45.000000000","submitted":"2019-04-23 01:02:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652865-1555981365995-5004bf0a","meta_rev_id":"35ed8f5d24b661ac1719c287bf3af0438b4fd85d","_number":652865,"virtual_id_number":652865,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-04-23 01:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-22 21:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-18 09:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-04-18 06:50:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-22 21:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 06:50:15.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-04-18 09:22:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-04-22 21:42:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-23 01:02:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07b003df80e703927de761d582e8cb151b884710","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:56:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d19c5c486cbb8d04723c8d9a3bd96af93d95e4f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:38:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/1/check/openstack-tox-docs/9cf73e5/html/ : SUCCESS in 7m 06s\n- openstack-tox-validate http://logs.openstack.org/65/652865/1/check/openstack-tox-validate/fbc4b46/ : FAILURE in 5m 04s\n- releases-tox-list-changes http://logs.openstack.org/65/652865/1/check/releases-tox-list-changes/63d56a7/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fe79d036c2937e9424c5fa18f9c1436f4eb99634","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-17 03:03:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f2edb177f9246e2b641f6c6eee0cf5cdb4c2c2e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 03:18:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/2/check/openstack-tox-docs/ff879d6/html/ : SUCCESS in 7m 16s\n- openstack-tox-validate http://logs.openstack.org/65/652865/2/check/openstack-tox-validate/26b2d69/ : FAILURE in 5m 30s\n- releases-tox-list-changes http://logs.openstack.org/65/652865/2/check/releases-tox-list-changes/23136da/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"61af75edfc099bb2b7263d6e76dfb36417ed1ad7","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-18 06:04:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b12af2d015096d3c3da10bdd805e0738e7534bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 06:28:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/2/check/openstack-tox-docs/d42bc4b/html/ : SUCCESS in 7m 56s\n- openstack-tox-validate http://logs.openstack.org/65/652865/2/check/openstack-tox-validate/3b760a5/ : FAILURE in 4m 36s\n- releases-tox-list-changes http://logs.openstack.org/65/652865/2/check/releases-tox-list-changes/42b93ab/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f143d7144e01ee01dfb748d1255fc55476810676","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-18 06:30:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"af83fe722e52638ab3fd83e69cda8189f82448d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 06:40:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/3/check/openstack-tox-docs/875cda7/html/ : SUCCESS in 6m 35s\n- openstack-tox-validate http://logs.openstack.org/65/652865/3/check/openstack-tox-validate/018cc6a/ : SUCCESS in 4m 00s\n- releases-tox-list-changes http://logs.openstack.org/65/652865/3/check/releases-tox-list-changes/28b2c01/ : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"ef1613ce81cd34bd809ede6fccb6babc67fb351d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-18 06:50:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9384ea978ba50fc42f85f4064aca2d6fc4c76cd8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-18 09:22:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ffe3b8ad748115c1608d5a8c5984b69a6681f0d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-22 21:42:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ad1f2c9055070f328c7b703e54abe8a790d39485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:42:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e898efb14712862e3fc21287a22e09c86dceb234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:59:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/3/gate/openstack-tox-docs/a31f3b8/ : POST_FAILURE in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/65/652865/3/gate/openstack-tox-validate/caac487/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a12536a3ce09c378760c731da309e938a79ab16f","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-04-23 00:34:51.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"263658607864b4e713d7dff823f81b8eb31764ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:48:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/3/check/openstack-tox-docs/13b0eac/html/ : SUCCESS in 7m 24s\n- openstack-tox-validate http://logs.openstack.org/65/652865/3/check/openstack-tox-validate/c3bb1b2/ : SUCCESS in 4m 49s\n- releases-tox-list-changes http://logs.openstack.org/65/652865/3/check/releases-tox-list-changes/b983887/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"bf8dccb2c8ad91bc257b5f214126fbed2b3636c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:48:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"03016d2cd27d07f20352f7b73fd5c1a8e0af32c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:02:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/652865/3/gate/openstack-tox-docs/8a4196e/html/ : SUCCESS in 8m 47s\n- openstack-tox-validate http://logs.openstack.org/65/652865/3/gate/openstack-tox-validate/a2d9f80/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"497701bf5d75b1b2c989dabaa84dbefa2377d2e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:02:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"53decddf73ca5038108cd1d758fdd97d1434fcc5","revisions":{"5cafc1e86c1c9b73eb747a757e5232ca5227de67":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:56:45.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/65/652865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/652865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/652865/1"}}},"commit":{"parents":[{"commit":"47e3d8b40ea4fc5bbef0ccdc3d4ea327bce87eb4","subject":"[murano] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47e3d8b40ea4fc5bbef0ccdc3d4ea327bce87eb4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[murano] create pike-em tag against final releases","message":"[murano] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I6bc59c8b97c852111c54be155c4ddb4f27cb9c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cafc1e86c1c9b73eb747a757e5232ca5227de67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5cafc1e86c1c9b73eb747a757e5232ca5227de67"}]},"branch":"refs/heads/master"},"89a8fee9507f243e2c9c871747459192c7aede63":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-17 03:03:56.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/65/652865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/652865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/652865/2"}}},"commit":{"parents":[{"commit":"74dfe50b6df1bbe596609a0642c13d041fb2b9ab","subject":"[murano] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/74dfe50b6df1bbe596609a0642c13d041fb2b9ab"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-04-17 03:03:56.000000000","tz":0},"subject":"[murano] create pike-em tag against final releases","message":"[murano] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I6bc59c8b97c852111c54be155c4ddb4f27cb9c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89a8fee9507f243e2c9c871747459192c7aede63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/89a8fee9507f243e2c9c871747459192c7aede63"}]},"branch":"refs/heads/master"},"53decddf73ca5038108cd1d758fdd97d1434fcc5":{"kind":"REWORK","_number":3,"created":"2019-04-18 06:30:31.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/65/652865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/652865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/652865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/652865/3"}}},"commit":{"parents":[{"commit":"74dfe50b6df1bbe596609a0642c13d041fb2b9ab","subject":"[murano] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/74dfe50b6df1bbe596609a0642c13d041fb2b9ab"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-04-18 06:29:59.000000000","tz":0},"subject":"[murano] create pike-em tag against final releases","message":"[murano] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I6bc59c8b97c852111c54be155c4ddb4f27cb9c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53decddf73ca5038108cd1d758fdd97d1434fcc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53decddf73ca5038108cd1d758fdd97d1434fcc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
