)]}'
{"id":"openstack%2Freleases~652867","triplet_id":"openstack%2Freleases~master~Ia2f17d020bb925e3c9e365baa3cf6e22d4f2ba58","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"Ia2f17d020bb925e3c9e365baa3cf6e22d4f2ba58","subject":"[mistral] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:56:56.000000000","updated":"2019-05-10 19:46:35.000000000","submitted":"2019-05-10 19:46:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652867-1557517595310-71816a1a","meta_rev_id":"e263ffe37b192a3f14acb68e8795c4217f4da2f4","_number":652867,"virtual_id_number":652867,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":2,"date":"2019-05-10 19:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-10 19:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 19:33:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 06:46:09.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-05-10 19:33:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 19:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecd319ce46777a54c3a6e19d6316311a44db1512","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:56:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbafc4a9687d96d64f79d739eb1abeab59f63bcc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-04-16 06:46:09.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"004a4c559d0ffdabe4596edc7a0fd9d7c4dbb729","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 06:55:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/652867/1/check/openstack-tox-docs/e9e1a8f/html/ : SUCCESS in 7m 20s\n- openstack-tox-validate http://logs.openstack.org/67/652867/1/check/openstack-tox-validate/99e3219/ : FAILURE in 6m 06s\n- releases-tox-list-changes http://logs.openstack.org/67/652867/1/check/releases-tox-list-changes/fcf4ed7/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ad0b18fcdef98954bab60e4466464ee60606097a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 05:40:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0b5c3ee2796e6f9bdbf2533e967ce93554520581","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 05:52:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/652867/2/check/openstack-tox-docs/3a20c90/html/ : SUCCESS in 7m 27s\n- openstack-tox-validate http://logs.openstack.org/67/652867/2/check/openstack-tox-validate/f48d5cf/ : FAILURE in 5m 07s\n- releases-tox-list-changes http://logs.openstack.org/67/652867/2/check/releases-tox-list-changes/c472975/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ea7d1c13e0717d6cc94538c5a2f92014d26585b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 18:30:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4d258c7486697d27941f20586a404a7e3ed9b5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:43:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/652867/3/check/openstack-tox-docs/d5ffa83/html/ : SUCCESS in 7m 28s\n- openstack-tox-validate http://logs.openstack.org/67/652867/3/check/openstack-tox-validate/b8881f3/ : SUCCESS in 4m 35s\n- releases-tox-list-changes http://logs.openstack.org/67/652867/3/check/releases-tox-list-changes/31650d0/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"93dbc8dfedfbd23b0bdeecb7a7329478e2f58a8a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 19:33:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c458a43cff577c67bc1a676f50b2498e71df7f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:33:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3326e83bb94192ed836e9518d9b26bba69198ebc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:46:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/652867/3/gate/openstack-tox-docs/0be9eff/html/ : SUCCESS in 7m 00s\n- openstack-tox-validate http://logs.openstack.org/67/652867/3/gate/openstack-tox-validate/597aa58/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"2c742333a744cc84a624faef1b6ff7a27682532f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:46:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01072817cc66e58922b59873700d941101aa43e1","revisions":{"c71decc54d898098642b1d553dd44e8b39c417ab":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:56:56.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/67/652867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/652867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/652867/1"}}},"commit":{"parents":[{"commit":"c40555d7cf3330974719dc331612acaaee075877","subject":"[mistral] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c40555d7cf3330974719dc331612acaaee075877"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[mistral] create pike-em tag against final releases","message":"[mistral] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia2f17d020bb925e3c9e365baa3cf6e22d4f2ba58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c71decc54d898098642b1d553dd44e8b39c417ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c71decc54d898098642b1d553dd44e8b39c417ab"}]},"branch":"refs/heads/master"},"357b7f6d3899c2b96b073fcaee0d2b4c83a35fc3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-17 05:40:01.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/67/652867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/652867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/652867/2"}}},"commit":{"parents":[{"commit":"508dacf41e70917c9025fd5a625f78d3d76ccb9e","subject":"[mistral] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/508dacf41e70917c9025fd5a625f78d3d76ccb9e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-17 05:32:40.000000000","tz":600},"subject":"[mistral] create pike-em tag against final releases","message":"[mistral] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia2f17d020bb925e3c9e365baa3cf6e22d4f2ba58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/357b7f6d3899c2b96b073fcaee0d2b4c83a35fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/357b7f6d3899c2b96b073fcaee0d2b4c83a35fc3"}]},"branch":"refs/heads/master"},"01072817cc66e58922b59873700d941101aa43e1":{"kind":"REWORK","_number":3,"created":"2019-05-10 18:30:48.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/652867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/67/652867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/67/652867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/67/652867/3"}}},"commit":{"parents":[{"commit":"3dbd750462b7d98ea855a81caa503a385c879ab3","subject":"Merge \"[heat] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3dbd750462b7d98ea855a81caa503a385c879ab3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 18:30:43.000000000","tz":-300},"subject":"[mistral] create pike-em tag against final releases","message":"[mistral] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: Ia2f17d020bb925e3c9e365baa3cf6e22d4f2ba58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01072817cc66e58922b59873700d941101aa43e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01072817cc66e58922b59873700d941101aa43e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
