)]}'
{"id":"openstack%2Freleases~652875","triplet_id":"openstack%2Freleases~master~I40cdfc99882f9a35b61e70df3c3de120fc81ebab","project":"openstack/releases","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I40cdfc99882f9a35b61e70df3c3de120fc81ebab","subject":"[trove] create pike-em tag against final releases","status":"MERGED","created":"2019-04-16 05:57:44.000000000","updated":"2019-05-10 19:46:34.000000000","submitted":"2019-05-10 19:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652875-1557517594329-5c776a74","meta_rev_id":"5c14a3ef1024471f5eb62a3dcca40d8b35bc0a45","_number":652875,"virtual_id_number":652875,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-10 19:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-10 19:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-10 19:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 05:57:44.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-05-10 19:31:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-10 19:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73e2d9628350c7e3275915df76a07795a3b7c3c0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-16 05:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07925a29ed8012adbc0c366f50de2eee289ae03e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 07:27:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/75/652875/1/check/openstack-tox-docs/a5d6d5e/html/ : SUCCESS in 7m 15s\n- openstack-tox-validate http://logs.openstack.org/75/652875/1/check/openstack-tox-validate/3715511/ : FAILURE in 4m 31s\n- releases-tox-list-changes http://logs.openstack.org/75/652875/1/check/releases-tox-list-changes/2cd390a/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0c4d43a8a3bab345a7458039299dfc6f83e20d7a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 16:17:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6383aaf243b1731065c4d12c82b2611377b867e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 16:17:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3bbce4def21b412097b95e14a56ed6800b0bdcfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:27:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/75/652875/2/check/openstack-tox-docs/6544315/html/ : SUCCESS in 7m 13s\n- openstack-tox-validate http://logs.openstack.org/75/652875/2/check/openstack-tox-validate/617e4f5/ : FAILURE in 4m 18s\n- releases-tox-list-changes http://logs.openstack.org/75/652875/2/check/releases-tox-list-changes/e8f4311/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"426b531e556d8baf657666f6af8aa5a4dc2508a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 18:33:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e5991b4b3348216f2309027d19ef8b298b4e781e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:45:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/75/652875/3/check/openstack-tox-docs/26201c4/html/ : SUCCESS in 6m 49s\n- openstack-tox-validate http://logs.openstack.org/75/652875/3/check/openstack-tox-validate/110fd38/ : SUCCESS in 3m 33s\n- releases-tox-list-changes http://logs.openstack.org/75/652875/3/check/releases-tox-list-changes/a154b54/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"3641eb2c37e00d02113800ae9b2be2251f4df4e5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-10 19:31:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c37b841d118140bfca18980e1163a2b24887b1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:31:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e170482fe3b7adb4cf29d02fe4f28a87ae77ab2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:46:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/75/652875/3/gate/openstack-tox-docs/a34eb38/html/ : SUCCESS in 8m 07s\n- openstack-tox-validate http://logs.openstack.org/75/652875/3/gate/openstack-tox-validate/13b8d28/ : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"1bc8427f042b42058c5802b91af99d56845243db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f908a184ce28661e3d609482907f31c5d8119a07","revisions":{"6aeac65d0339ff5bf9f1295bc9def8de1988357d":{"kind":"REWORK","_number":1,"created":"2019-04-16 05:57:44.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/75/652875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/652875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/652875/1"}}},"commit":{"parents":[{"commit":"7816e85986594d3064666d178926d572ac4020c0","subject":"[trove] final releases for pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7816e85986594d3064666d178926d572ac4020c0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"subject":"[trove] create pike-em tag against final releases","message":"[trove] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I40cdfc99882f9a35b61e70df3c3de120fc81ebab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6aeac65d0339ff5bf9f1295bc9def8de1988357d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6aeac65d0339ff5bf9f1295bc9def8de1988357d"}]},"branch":"refs/heads/master"},"bc429dd81f2876c3b7346dec96f76c582a3ff6cc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-10 16:17:30.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/652875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/652875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/652875/2"}}},"commit":{"parents":[{"commit":"3813b476ed01fab9e5eb2283d7fe7b895573b7a9","subject":"Merge \"[trove] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3813b476ed01fab9e5eb2283d7fe7b895573b7a9"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 16:17:30.000000000","tz":0},"subject":"[trove] create pike-em tag against final releases","message":"[trove] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I40cdfc99882f9a35b61e70df3c3de120fc81ebab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc429dd81f2876c3b7346dec96f76c582a3ff6cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bc429dd81f2876c3b7346dec96f76c582a3ff6cc"}]},"branch":"refs/heads/master"},"f908a184ce28661e3d609482907f31c5d8119a07":{"kind":"REWORK","_number":3,"created":"2019-05-10 18:33:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/652875/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/652875/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/652875/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/652875/3"}}},"commit":{"parents":[{"commit":"3813b476ed01fab9e5eb2283d7fe7b895573b7a9","subject":"Merge \"[trove] final releases for pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3813b476ed01fab9e5eb2283d7fe7b895573b7a9"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-04-16 05:53:18.000000000","tz":600},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-10 18:33:12.000000000","tz":-300},"subject":"[trove] create pike-em tag against final releases","message":"[trove] create pike-em tag against final releases\n\nThis is a procedural review to transition pike into extended maintenance by\ncreating the pike-em tag\n\nIt\u0027s possible that the release isn\u0027t needed (for example if there are no code\nchanges sine the previous release, sadly that\u0027s non-trivial to detect en masse.\n\nIf you are the PTL or release liaison for this team please +1 this review\nif it is ready or -1 and take over the review to correct mistakes.\n\nChange-Id: I40cdfc99882f9a35b61e70df3c3de120fc81ebab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f908a184ce28661e3d609482907f31c5d8119a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f908a184ce28661e3d609482907f31c5d8119a07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
